)]}'
{"id":"openstack%2Fmistral~809818","triplet_id":"openstack%2Fmistral~stable%2Fxena~I461b13377aec7b00cdc7837aec5c87334c5d8d6b","project":"openstack/mistral","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I461b13377aec7b00cdc7837aec5c87334c5d8d6b","subject":"Update .gitreview for stable/xena","status":"MERGED","created":"2021-09-17 16:22:32.000000000","updated":"2021-10-04 21:52:07.000000000","submitted":"2021-10-04 21:50:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809818-create-xena","meta_rev_id":"3b6bafef30866b3ecdf7e7b4fbddc0e368bae433","_number":809818,"virtual_id_number":809818,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 21:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-09-23 10:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-30 17:50:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-10-01 16:28:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 17:13:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-18 16:09:18.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"CC"},{"updated":"2021-09-23 10:10:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-30 17:50:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"f57f475cb705c9fa7f198388cc723791f95a53b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-17 16:22:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11a96bd6387fce92ff0927408ae799fb143bce3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:13:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0d8a0816534a84b94bc79df259b6f3 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b294dbba91464a4aad4c00b1f071dab7 : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5b94bdf7ab940faa66fe52a5cbbc7b3 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1518b336800f497fba1e778d4d78b7a8 : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa911e2f1251494facc87c423d161a40 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd807c9540394984bfe6b9ff6b7c10d2 : SUCCESS in 9m 19s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/57131e29fc134e1bbbe4ebea09e7b83c : FAILURE in 31m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/26680ae23e114dbb95002318b42e7351 : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a4d03c983d917b19b2300bbc669704078861c48c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-09-18 16:09:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93a8bc3be681080de001e3b64ad80f906aea3b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 16:43:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70be1e64ca5400bb0fc36155ac7dae6 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c6c302d276e4fd5b59881da13bd34d3 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/642c0caf21694313a077c73a87a1df25 : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16434d90214d45aa94be1dba9ecaeb6b : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70b25ad1270487ebc42ca8e3d8e556c : SUCCESS in 7m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3623466479453cb8f00240a00eaf61 : SUCCESS in 9m 41s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f8c3213ca8cf401b9914a2af6a4e3d41 : FAILURE in 29m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c37dc18d2bac46ecaf6aae1a6a46c765 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fbb04fe9b6ee3cdca1f6b0f4bed4aa47151fadca","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-09-18 18:18:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"468c4c087658004609b605900d5980cbad47210f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 18:49:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafee3f2e2b84ed99c41e9227e262323 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11f8298484404f9e90007b99a473fb17 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/303e33990959495e8cfa8f8964ce7854 : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab24903e567f4b9e94ab88535449c47e : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e324168ea9ed43dbb346f724140c19d7 : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc9d4b72cfe34fa39b22f734cc8621c0 : SUCCESS in 10m 16s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f1472e7610a64f6385665736392092c7 : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/958bc14137d54a86a988ae9ba7d8bdd3 : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"77d25c7d0e2ffdfd9c378e4141e219380f0aab32","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-23 10:10:20.000000000","message":"Patch Set 1: Code-Review+2\n\nGenerated patch, looks OK.","accounts_in_message":[],"_revision_number":1},{"id":"d10b087b45b8ed1ff78b2a167358cc1e7f643300","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-30 17:50:14.000000000","message":"Patch Set 1: Code-Review+1\n\nlet\u0027s merge this to setup stable/xena","accounts_in_message":[],"_revision_number":1},{"id":"5e49b5c3a878d1eb40d5ea31402441c4dc9543e7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-01 16:28:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f317fceef861e7e656d9fa358e2ca462237d4b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 16:35:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29c3380b09c55f1f5471f0bee7f09d7c4f1f8260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:24:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1a02f2a33348efba86a34d3a43c8c6 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6006c16b6ca14b6481da5360ccf65810 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abb29cdb56194759b84d1dba969b1159 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02e556df3704a6bba7370c9f8d7bbc4 : SUCCESS in 12m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9f76eb19a73d42b1ba2402703cf60816 : SUCCESS in 41m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3715a9e54a0647b8aa380f615551a3a3 : SUCCESS in 42m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4043639804ac45b683c5256523c44f19 : SUCCESS in 44m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/122bbcbb9e394c9e8a47f1650ee7c890 : SUCCESS in 41m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9c686ba2ba8e46279e062f147c9c9a6b : SUCCESS in 41m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8a45c6a95fdb414fbf1eedd8c8cd4322 : FAILURE in 29m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1e12ca877da74c4c9ab89a0b944d1a06 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8fde88cd94d67ba18d2fd95af2404b1177001e4c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-04 14:25:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc7cad088e7a37ffeb7a76b2147278ed99282a9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 20:44:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89df4efed1104433a3def9d5df5652b0 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c27ffd842ccc40fab7ae80c3e8abebb3 : SUCCESS in 9m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f10aeb1b0bdc470f81081c9e8fefcd47 : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa4361fbb363480ea6c786518286c7fd : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc751f6ff5944ccabcfa24bb729b3df : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/567c5519488d486c818d3170bb721cc8 : SUCCESS in 10m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b409eff8c7ef4cf2a29416a9e4e59e50 : SUCCESS in 43m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/636c0f00c93c4fc38ecc225e9309b8f8 : SUCCESS in 41m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3f8527a6f6f42e8bb3ac287a00c8b24 : SUCCESS in 40m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d72db8f28d5d4350ac516b916bbe9006 : SUCCESS in 42m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b861fb3936d244fd912a907335529c8d : SUCCESS in 41m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ded3d866519b4098a0b4500136b1d451 : SUCCESS in 30m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/57a020411a754e1abdd06e08587c2e15 : SUCCESS in 29m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/217829172f6346028c0ce8c38e243aa6 : RETRY_LIMIT in 2h 02m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e27868e2d02d884fc5b52d106d47386e9805686f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 20:46:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b10691c42826b0b01f3e35e4804d058c5ebaee99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 21:50:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f044e56b0e444d3a93ff58e4d7419d41 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a063a0c717e54830a33dcf62b4743138 : SUCCESS in 10m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05946d4a24794118ae529ed2d9c81816 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e2cc6a12a5465b8eccd1c464b3b129 : SUCCESS in 9m 05s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/066b8fbdcc2c42a3b071dd4672865ad5 : SUCCESS in 1h 02m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7696c7615d946068faaea0ff42caa66 : SUCCESS in 42m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bec394894b5142839aa01bef3cbcd2af : SUCCESS in 38m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/36472f6d99474834860459aeb23bb3f8 : SUCCESS in 45m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eb7e1d07ede2453f9064b2e9ba3730c5 : SUCCESS in 41m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c570e865a5d74afe94436b69a6375b53 : SUCCESS in 29m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3a562d8fb0a349b596368154a666e11a : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":1},{"id":"97507c730677248dba44afb5a3b666052cf1018b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 21:50:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b6bafef30866b3ecdf7e7b4fbddc0e368bae433","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 21:52:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c2ddf606b646668e71d127bae3c931 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe11ac57adab77e459f6fb73689aa8b48395c5e6","revisions":{"fe11ac57adab77e459f6fb73689aa8b48395c5e6":{"kind":"REWORK","_number":1,"created":"2021-09-17 16:22:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/18/809818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/809818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/809818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/809818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/809818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/809818/1"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:22:31.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:22:31.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I461b13377aec7b00cdc7837aec5c87334c5d8d6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fe11ac57adab77e459f6fb73689aa8b48395c5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fe11ac57adab77e459f6fb73689aa8b48395c5e6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
