)]}'
{"id":"openstack%2Fmistral~809819","triplet_id":"openstack%2Fmistral~stable%2Fxena~I4a4cacc358fe7700347fc3f2c2b4ae44a4b60214","project":"openstack/mistral","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I4a4cacc358fe7700347fc3f2c2b4ae44a4b60214","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","status":"ABANDONED","created":"2021-09-17 16:22:37.000000000","updated":"2024-03-26 18:36:13.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d0a808dcf035539ea2e109d6b882ee6f7b475304","_number":809819,"virtual_id_number":809819,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-16 15:20:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-09-23 10:10:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 17:50:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-10-01 16:28:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 17:12:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-18 16:10:10.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"CC"},{"updated":"2021-09-23 10:10:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-30 17:50:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"9285a7b17837044d02bcd6129219ae0bf5282429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-17 16:22:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca2e8bf6e5d200afbeecacc9b4bd016f10067213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:12:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2d62bc97254b4f85b2d8e2d8cb49ca : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88fa3dc5a5d84ccb84d02b3457799832 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe9ea1d310b4475c803734ce70d6ecec : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1318e0c8e5b84fbca3b33d9f2c65ba16 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cf517ac201467f9ebcd8afc2b6dcac : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e49b8b360af435a8bf7ccd9b72239e5 : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51eceff0f3484035bd75b2980e6062f1 : SUCCESS in 13m 11s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3a4b8594d9924077b212aaabf8834166 : FAILURE in 33m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3ed3930c52864090a014cdc2c86e3293 : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"eab8b54fdb9bd2b4716abc0020e9ecba1ba3a95a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-09-18 16:10:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65d26bcdfa8ae9862a98d39b1d133cd1f2fc9c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 16:43:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae16346bf0644ec6b3f58532893e26a8 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b2fdc978a3d4587803e160b2a1000d4 : SUCCESS in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d084f88b784f72b817a38bb9de641e : FAILURE in 9m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a73830e254b40cb857012ff18bf1b85 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe193c36a36d4d4ca9095abe4ad36921 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/277faaa8960c41e9ad428e966b20e580 : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bdd5a0166e44e4b838c2a0ab844c26c : SUCCESS in 9m 37s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9f780c31558e4ce3b90da9c026dda550 : SUCCESS in 29m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a5a4517f32ac45d6b52e26c0fb8998e0 : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d94cd1186fe9a32f12b393c66561ef0f432a5553","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-09-18 16:45:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"68b4e4557a4f0d700ec4584dab57e3438959b2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 17:18:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d76c6d6c2934af89bac91d0e788e28b : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a945e056ab64eecb3ffadbe88dc1838 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26720c0e6c0c4b788da3ddce62837fba : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/819bd8b5121142a495eff6e3445fc8e9 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418fb95d05e84b57b96818d7b5a0f82c : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ee7dd6f0a9d4542a06812f0f91bc5cf : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cb60ac61ecf4f3592a1a2ef67d232ea : SUCCESS in 10m 29s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3c356dc19a184e5d8c08a3195c00d6e2 : SUCCESS in 31m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f58a7ee62f2f4d2f8234c77192093408 : SUCCESS in 28m 59s","accounts_in_message":[],"_revision_number":1},{"id":"14b8f7862362d1bc103528771ca7631c7fc0bb03","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-23 10:10:34.000000000","message":"Patch Set 1: Code-Review+2\n\nGenerated patch, looks OK.","accounts_in_message":[],"_revision_number":1},{"id":"884cbc801bf873e41c6265411382174fdc2f1eed","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-30 17:50:18.000000000","message":"Patch Set 1: Code-Review+1\n\nlet\u0027s merge this to setup stable/xena","accounts_in_message":[],"_revision_number":1},{"id":"1f32c0c5da80a617343e7a004cde21d126587e15","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-10-01 16:28:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1bce825d58ce368e5904fa623e48cf91cff3c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 16:35:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"546b813766cd74d79b30bf46e2fa9bcf69ea4486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:24:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"7582706185758cd27a5ae88eb065ad608bd38d2f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-16 13:12:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"466fe51267c6a5a41a3a10ea644662d4053e2d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 15:20:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d36c052513b4e638604166aa582e42e : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/532106cb51ea4ab8b72b6dfd29bef92c : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90e8f72e54b9411d84dbc4da015ef5b3 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b230fad1b75745558060af62e366f2d7 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3f83e4964a490bbc308fe18e39687c : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea8831adb4544828b362ba202c9c149a : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d07d4f4d5e4a46efa790e83573298530 : SUCCESS in 9m 34s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0dc056c183d34d6ab3dd9e25c9e4456f : SUCCESS in 36m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2084816d83fe41e7bacaba995a6454ea : SUCCESS in 50m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e9d4726a2294626b22d2d20ea370054 : SUCCESS in 33m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ff7b06cc88d64201920884ef97f68126 : SUCCESS in 40m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cd35ce628cb04cd1a682d14f032a41d4 : SUCCESS in 34m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3662b500d4744f84b1090d93e8e4bef2 : FAILURE in 28m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9cd65fa68a274a4a98654f1afa290e20 : SUCCESS in 28m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/68ed4502ff8e4fbe956b54d65901ad73 : TIMED_OUT in 2h 05m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0a808dcf035539ea2e109d6b882ee6f7b475304","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:36:13.000000000","message":"Abandoned\n\nstable/xena branch of openstack/mistral is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/xena if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b396834a18d35210d7356b9f8be237592893945","revisions":{"7b396834a18d35210d7356b9f8be237592893945":{"kind":"REWORK","_number":1,"created":"2021-09-17 16:22:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/19/809819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/19/809819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/19/809819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/19/809819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/19/809819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/19/809819/1"}}},"commit":{"parents":[{"commit":"fe11ac57adab77e459f6fb73689aa8b48395c5e6","subject":"Update .gitreview for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fe11ac57adab77e459f6fb73689aa8b48395c5e6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:22:32.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:22:32.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","message":"Update TOX_CONSTRAINTS_FILE for stable/xena\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/xena branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I4a4cacc358fe7700347fc3f2c2b4ae44a4b60214\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b396834a18d35210d7356b9f8be237592893945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b396834a18d35210d7356b9f8be237592893945"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
