)]}'
{"id":"openstack%2Fmistral~821437","triplet_id":"openstack%2Fmistral~master~Ib9dd961ae10adc467233bdb6bef2fe1a89778f01","project":"openstack/mistral","branch":"master","topic":"yaml-load","attention_set":{},"removed_from_attention_set":{"11805":{"account":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"last_update":"2023-12-13 23:04:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib9dd961ae10adc467233bdb6bef2fe1a89778f01","subject":"Switch test_workflows.py to yaml.safe_load","status":"ABANDONED","created":"2021-12-10 18:08:41.000000000","updated":"2023-12-13 23:04:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"89f3d55cee60846bec202e4ef146602682bb74bb","_number":821437,"virtual_id_number":821437,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-10 18:49:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 18:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a73bba007b4e4043b1fe2d2338d1054979a79ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2021-12-10 18:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb9f5b13f24175175cb046f33b971f30ec1cca02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2021-12-10 18:14:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"91c99a559198557166f23e24c14f3430bb647390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 18:49:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cab3c7e85c4ed292c3df35f004f753 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4dcc7a141f4d48956e90f63aaa171d : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a71e9e1d23ed42c4bdc5a5df400b3b26 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eefc873aef2b4d78ab69a9b18878df08 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1973b2931084443d87e162dd1b95ff9d : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5788c49fdd0f45c098678d3c388f42f7 : SUCCESS in 8m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/28f8fe80e2984d18996002d30079c231 : FAILURE in 17m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c960d1c882cc4fe2a1e6a5fcc340ab5a : FAILURE in 21m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88b3f912c400410cafa3e5cb613b11eb : FAILURE in 15m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/788eb2348d844e9b8362f339acc65dd9 : FAILURE in 27m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/18ef05511d2a45dba4ed88426aee50df : FAILURE in 28m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2d55eb07f9324d4fa2150a334f02123d : SUCCESS in 30m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2c4b88d779e74c31b683f774a6e2fdf6 : SUCCESS in 28m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8f9e564df3194465950e633d7e1c891d : FAILURE in 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89f3d55cee60846bec202e4ef146602682bb74bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-13 23:04:07.000000000","message":"Abandoned\n\nfixed with https://review.opendev.org/c/openstack/mistral/+/801223","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"74cef08fbed28887158d739e9e3cd2b8082221c0","revisions":{"98df99320d2dff79677b4eb0adaef85e5ff3e29e":{"kind":"REWORK","_number":1,"created":"2021-12-10 18:08:41.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/37/821437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/37/821437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/821437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/821437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/821437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/37/821437/1"}}},"commit":{"parents":[{"commit":"36d25829612b00fbd8f41a4ab060a851c4a69e67","subject":"Switch to collections.abc abstract base classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/36d25829612b00fbd8f41a4ab060a851c4a69e67"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2021-12-10 18:01:45.000000000","tz":-300},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2021-12-10 18:08:20.000000000","tz":-300},"subject":"Switch test_workflows.py to yaml.safe_load","message":"Switch test_workflows.py to yaml.safe_load\n\nNew versions of PyYAML require the Loader argument.\nAnother option is to just call safe_load() to load\nthe YAML safely with yaml.SafeLoader.\n\nCloses-Bug: #1954475\nChange-Id: Ib9dd961ae10adc467233bdb6bef2fe1a89778f01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/98df99320d2dff79677b4eb0adaef85e5ff3e29e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/98df99320d2dff79677b4eb0adaef85e5ff3e29e"}]},"branch":"refs/heads/master"},"74cef08fbed28887158d739e9e3cd2b8082221c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-10 18:14:55.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/37/821437/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/37/821437/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/821437/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/821437/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/821437/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/37/821437/2"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2021-12-10 18:01:45.000000000","tz":-300},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2021-12-10 18:13:57.000000000","tz":-300},"subject":"Switch test_workflows.py to yaml.safe_load","message":"Switch test_workflows.py to yaml.safe_load\n\nNew versions of PyYAML require the Loader argument.\nAnother option is to just call safe_load() to load\nthe YAML safely with yaml.SafeLoader.\n\nCloses-Bug: #1954475\nChange-Id: Ib9dd961ae10adc467233bdb6bef2fe1a89778f01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74cef08fbed28887158d739e9e3cd2b8082221c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/74cef08fbed28887158d739e9e3cd2b8082221c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
