)]}'
{"id":"openstack%2Fmistral~822645","triplet_id":"openstack%2Fmistral~master~I4d3f45c54c5381e1f110ae993e0f147147f3a891","project":"openstack/mistral","branch":"master","topic":"fix-collections","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-06-20 11:14:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4d3f45c54c5381e1f110ae993e0f147147f3a891","subject":"Fix import from collections","status":"ABANDONED","created":"2021-12-22 10:22:05.000000000","updated":"2023-06-20 11:14:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe450e7b95164cc953433ad56b8c081829bf812e","_number":822645,"virtual_id_number":822645,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-22 11:01:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 11:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc38de1c7cab60e260eea39555005e3f70341177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-12-22 10:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"031c8b493d164d06e79fff73ae2da14d6700cb4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 11:01:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13995cf6827456992a0fbadcc2b1237 : FAILURE in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf8850fff3cc47caae046cbb93d03600 : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46655dc8549844b883b1b0dee3e3a492 : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6e3cd0dd13b4a96b8124d185020dee8 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5796adba9f484fdfa6a2fb7911ad153c : SUCCESS in 7m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f43639fd40f4d3690ff40ad6854d743 : FAILURE in 5m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1f9742668adf49b28108785c0eaed848 : FAILURE in 30m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/783ceb0e57fa48889f0e734fbfe1a923 : FAILURE in 32m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a3bf3a4f6b431b82ff6d6d17e21461 : FAILURE in 27m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/51bf7d12f980469c8cbd7ff5ebc8ffba : FAILURE in 23m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b03bd8c65123430cbe329e629d015cc0 : FAILURE in 25m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/696b452ccdc747108ce16c64785720d9 : FAILURE in 28m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d8e27d4ed49441da8e173b6b2f177c8a : FAILURE in 29m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dc79fee6686c4eb7877fece3bedad6ef : FAILURE in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe450e7b95164cc953433ad56b8c081829bf812e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-06-20 11:14:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4997879514270cc87cf708d8d9a720fe0d3f68c0","revisions":{"4997879514270cc87cf708d8d9a720fe0d3f68c0":{"kind":"REWORK","_number":1,"created":"2021-12-22 10:22:05.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/45/822645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/822645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/822645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/822645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/822645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/822645/1"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-22 10:21:08.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-22 10:22:02.000000000","tz":60},"subject":"Fix import from collections","message":"Fix import from collections\n\nImporting Iterator and Iterable from collections has been deprecated for\nyears, and is now failing under Python 3.10. This patch fixes the path.\n\nChange-Id: I4d3f45c54c5381e1f110ae993e0f147147f3a891\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4997879514270cc87cf708d8d9a720fe0d3f68c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4997879514270cc87cf708d8d9a720fe0d3f68c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
