)]}'
{"id":"openstack%2Fmistral~822876","triplet_id":"openstack%2Fmistral~master~Ie403fcac5cfd3a02f471b6b5eda607392b363381","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-08-31 00:58:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie403fcac5cfd3a02f471b6b5eda607392b363381","subject":"Update python testing classifier","status":"ABANDONED","created":"2021-12-24 06:19:35.000000000","updated":"2022-08-31 00:58:46.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9167bf3d810a5711ffd241053c45af80c7ac02c5","_number":822876,"virtual_id_number":822876,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-24 06:50:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2022-02-03 15:42:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 06:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 15:42:22.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"},{"updated":"2022-02-03 15:42:33.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"f8365a2d8294256d3f494ac1d4f6d97ba4d7b3ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2021-12-24 06:19:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abe217c3adc3a621bd6da22f9db538ba2f896f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 06:50:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a1fb0b7fd5443609f8d34938439c513 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9be85880d7c481ba95338c447a3a5fd : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acc188d366b44ff89809f621e4919536 : FAILURE in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb4927957e8495ba74fdefc69a15980 : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/002862047c504475a0519e2a8d1e99a5 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91ffae90a84441f99e0e587e9d21fc6 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbdb0e2a320f450ab05c74a64f4f4051 : FAILURE in 3m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0faa6b19b7e14f349ca1845bc48df196 : FAILURE in 18m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7b5a6c3a25472480f86772acf5a124 : FAILURE in 20m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5abcc9ee09b4b39838e8398cd9e197a : FAILURE in 19m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a89d0bfcb27b4fb9bedde370cb6d7740 : FAILURE in 22m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c143a8b665224e92933b8658fe7ec606 : FAILURE in 19m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/965029b9b2db4f8d92a029ae7e35204a : FAILURE in 29m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e707694cb58c4979a11da46775446bb3 : FAILURE in 26m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/83aecd5a3189430d860be9567f2474e4 : FAILURE in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f180ddd6e6c54866aa359bcce35d95f272ed3cd","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-02-03 15:42:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6d573b9a10933c4534d9fcbff76148a71d14cb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-02-03 15:42:33.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"9167bf3d810a5711ffd241053c45af80c7ac02c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-08-31 00:58:46.000000000","message":"Abandoned\n\nIt is duplicate of https://review.opendev.org/c/openstack/mistral/+/822467","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad009897867ab7da2b00da5d99f9f0650800c3aa","revisions":{"ad009897867ab7da2b00da5d99f9f0650800c3aa":{"kind":"REWORK","_number":1,"created":"2021-12-24 06:19:35.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/76/822876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/822876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/822876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/822876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/822876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/822876/1"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-24 06:17:56.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-24 06:17:56.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: Ie403fcac5cfd3a02f471b6b5eda607392b363381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad009897867ab7da2b00da5d99f9f0650800c3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad009897867ab7da2b00da5d99f9f0650800c3aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
