)]}'
{"id":"openstack%2Fmistral~826425","triplet_id":"openstack%2Fmistral~master~Ief9d24e83d74c1f7bf17c96abd713ee5b563361d","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"34340":{"account":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"last_update":"2022-08-12 18:14:46.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-08-12 17:11:40.000000000","reason":"\u003cGERRIT_ACCOUNT_19134\u003e replied on the change","reason_account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}},"hashtags":[],"change_id":"Ief9d24e83d74c1f7bf17c96abd713ee5b563361d","subject":"remove unicode from code","status":"MERGED","created":"2022-01-26 08:11:35.000000000","updated":"2022-08-12 18:16:39.000000000","submitted":"2022-08-12 18:14:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826425","meta_rev_id":"d3d0526b0406dbc0305ca93695587e6ec5dc4a1f","_number":826425,"virtual_id_number":826425,"owner":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 18:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-12 17:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-12 17:11:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 08:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 10:10:09.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"926b16788dfac4759e334b36ddf73a75b9b2aaea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-01-26 08:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8b6708f9aa8df871442c6372bcee3a18b7ed318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 08:47:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e09466a75f46aea829bcc9bdb6df2f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdb8a90b99d24e4483ce9b140366e4a9 : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccd3a8bb4018472f85312a416bdb69c4 : FAILURE in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c33559401c1461ab2fb1c018641c2dc : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd838a71cd64b4e93e63c659aa615b8 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcbd2bc1aaf448e385970ef2632ea5f0 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca7d535caf37402d86a9118f348a6d73 : FAILURE in 5m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e66d43f314954cb590a5cc4069ec3262 : FAILURE in 26m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09d38bfd4e0c44c69831c86816c5e743 : FAILURE in 19m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0314e22c2fc74306b8f4fca2fcd2d551 : FAILURE in 12m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/71470180aa974186bb801de51e2238d7 : FAILURE in 27m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/755c5f2bc7aa430881fa34deb2403326 : FAILURE in 26m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bc3527e8ee404b50b60ebf9eee8ffe2c : FAILURE in 28m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/075d03f6d28746d082036d2ae7cf640f : FAILURE in 27m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3d61fbaa10a046189d1a3c0d06ef1e9e : FAILURE in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f275f18c14524f84a31c990f53b668274575f10","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-01-26 09:23:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a62faa9e92b1a50299366732c29161586fb77251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 09:58:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/827eef1b96244257be857dcd85ede35f : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8609cdf4101f4fed98b05605abb9c7e6 : FAILURE in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093bb018661849dfa660b7e9be21eef7 : FAILURE in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf332cde72b4458c83c81af9506f55ea : FAILURE in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058c83e1299e430aa494c1ea7ef0060c : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a9785d805e447429294ea0a4c8f93dc : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1993dc77224f44988b5df16e94733362 : FAILURE in 4m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/10c7fc0089ca4955b4f6d6091918e5e9 : FAILURE in 23m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afa2cdfb9f284004a2c2ce0f52e17beb : FAILURE in 23m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ce6b19251d045269fbe479f26c7b08b : FAILURE in 23m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/51be0a0d937d43daa9d9347fd3aca0db : FAILURE in 23m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ee635f75c27f450c81d78c97a5eb0df3 : FAILURE in 23m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/92f6c4b93e1348409a9ec2695c84a29b : FAILURE in 29m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5290bb0f2a884d79867de3337bfd84c0 : FAILURE in 26m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b47e920647034249afcf8764fc5d76d7 : FAILURE in 22m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5e56a992b05dc2ef90acdb19cdaa935472396e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-01-27 01:13:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a79424cd5fbf8f411ff57b115030a6210f569928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 01:41:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c56583d6a314ea2bb07983011055f97 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce328516fd4493f8c47258e95c53140 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e86a2613f84ba0a574c2cf5dc4661b : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6672e7f5ca34b9e8482c83e089719bd : FAILURE in 4m 26s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dd3623d65a86497689cdbff8618b8e08 : FAILURE in 27m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b5db5427ac1d450ebfc643625798d161 : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c26933fc935f32522ab679efae86c5ef8e736b87","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-02 10:10:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7eab713a43225d6d2d0f5c52d7b8d0a82d2946db","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-03-11 06:02:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb310cc1562401ab0295cc075e36d384f07c9cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 06:34:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da10a733cf6143a8bd96dd7763a3868e : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5579396bd938422cb0f9f08ac885ce4a : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5f7295323714ce28de2f2ca4c536fd0 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0665f0c2303f4ce9872550506a0da288 : SUCCESS in 10m 55s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0d5f9529164b40829cdb6c2b7468137a : SUCCESS in 30m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e1a89fe1d02d4223ba0f8dd01e3fb9d7 : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":2},{"id":"60d2bfcf5d0ea655a4855ca7f9270bf43eb9871f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 09:39:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c24b99f6ee648963d30d35e301d3f822a83bd2dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 09:40:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8da70d60fb9714ea77bbd0f7c12ef7ed2aec059f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 10:17:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7e83e5d3fd541dc84bcdd6f554b5edc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79017d98cafc4c42a68f0f2aecd302d6 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052977933a1b4eed945141626922cb98 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d02a8123651411c9b421b9789d9b05f : SUCCESS in 6m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/902bd0af10a94437bba5ebf3c16af47b : FAILURE in 32m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d00ced1934654b34b31595ac0ec62d5d : FAILURE in 30m 36s","accounts_in_message":[],"_revision_number":2},{"id":"5dd2e65105f3bd27e930205eae23337884eb0f22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 10:29:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b3ecad52141049e8958239e0f340cef392172ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 12:40:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e0bc727fb7b4bed806c8f4a5ce90f3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60390ae91d0491a94ca3d4fb0b4f930 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43b48d53f3b748a1bbf4fb8f488465cf : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2f39e30242d4763a876657af7cf6dd6 : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb0f035ba96e4baebfa7d14fc12afef4 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463043dc164643e8be239e2ed346166f : SUCCESS in 7m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fdd0760ca66d4b1cb4277f35687e4c8e : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25911b6907a047f6953a52127fc79864 : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a21def7c2f04cfeb608290b674266d5 : SUCCESS in 11m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bba4db1cb3c947bab0cab2fe8ab5e376 : SUCCESS in 55m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e42efb49d86f4f69ab0a64650bde8941 : SUCCESS in 51m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1666f53f6656461faa767f52b7002448 : SUCCESS in 49m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9ead416eb94b4943852a9633b569021f : SUCCESS in 48m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1f8b88fa6e574ee4bb21563f64f22b80 : SUCCESS in 53m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fedfcdd74fb841e29e92e507a96656de : SUCCESS in 34m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f2795a4c8e6e41099543912977a5986c : SUCCESS in 32m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4940b91396004205a78a954bcc8271f5 : FAILURE in 2h 03m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7c3c91ddb335ed0b8dbf4fe5b68b9900b3eab994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 14:41:59.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e38f6107dc147e6e7231480e60810ee921259e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 16:39:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecf19918b2f44dcd8f62c144c9e3cdbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e055dd18424b77acc50fca0eedf502 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4034efa884c748bba2d74fd11d96d261 : SUCCESS in 9m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/840185dabe414fcd8c0f1b0439074f45 : SUCCESS in 9m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/219a139335c744998808a9a70ff39e2a : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bd3ea4e02a4a55896a60dd25ed7c41 : SUCCESS in 7m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/08320301165d418a885671022b6ed598 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/269d410271514c148ed2cb2869272ce5 : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c344a8aacf7c4355a2b04170d2c8236b : SUCCESS in 11m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7f315eaf1e2b47e88a07fe0430aede21 : SUCCESS in 49m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93cf527fb1da4da99fe7576ff4a1b9e2 : SUCCESS in 50m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47453cac8b1743289d0211e8458e1e70 : SUCCESS in 53m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7829899c53484d47ae203c051dd17524 : SUCCESS in 50m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b7269a8b72bf419a9d71ec2931f1d9a4 : SUCCESS in 49m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7990a7f044174e738fd7935e09d0caa7 : SUCCESS in 35m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5556160ec53b4d679a67e79ff41a2580 : SUCCESS in 32m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/db34a8e9bcf448fbbb365bc33adbcf6e : FAILURE in 1h 51m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"77b7b18767798c8929bf11df6c85fea90d38bfb3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 17:11:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9b9ddb8ebdd9a6d428ed731a8b4bd74ac926b92b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 17:12:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4ef9eac198c00dd0f07a149a2fc85e01bcb44f3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 18:14:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2b053357004348924ad95a59561b9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7de13ab133d4d90a6472e8651c2156a : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df6c80f694de498cbb4d4e4a8d90e6dc : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f20d1773907451c9b1ebf5353e587e2 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ae7097a4564263a817725542126df1 : SUCCESS in 7m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef830f37ef624887bda0e01c48a2674e : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f44b557042445d48e70a18695692be3 : SUCCESS in 4m 43s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8bf7275a3b894da190bf1a1d57d78414 : SUCCESS in 48m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/879f4de8e614499a8993756e7128e501 : SUCCESS in 52m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75bae5c4fd214efa887e19aae49864d7 : SUCCESS in 48m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2cfa56e1e01d4805a13a81e1d4a0cc37 : SUCCESS in 51m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f5a621c19dbf4af4980247638477da28 : SUCCESS in 56m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c3406e95837b4151b3a16ce4115362ad : SUCCESS in 35m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b7574ad8f9bf415d9ce3d5051f10df5b : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":4},{"id":"a9c6c9e441f02bac30cb900b84ff0482763e0db3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 18:14:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d3d0526b0406dbc0305ca93695587e6ec5dc4a1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 18:16:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d282a9891047668181debd7cf9e597\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d180751bfdc049e49739e6316745e78c : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a52e707b23345238fee89887133fdeb : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fbf80913644787e283b058eecd1ffd3aa9eb01b4","revisions":{"ff55f53428ecd3e191f34f00d2998e8a082e3774":{"kind":"REWORK","_number":1,"created":"2022-01-26 08:11:35.000000000","uploader":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"ref":"refs/changes/25/826425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/25/826425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/25/826425/1"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-01-26 08:11:00.000000000","tz":480},"committer":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-01-26 08:11:00.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nremove unicode from code\n\nChange-Id: Ief9d24e83d74c1f7bf17c96abd713ee5b563361d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff55f53428ecd3e191f34f00d2998e8a082e3774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ff55f53428ecd3e191f34f00d2998e8a082e3774"}]},"branch":"refs/heads/master"},"59858915222ce7b2670ea0226fcf8a50a2ea7fac":{"kind":"REWORK","_number":2,"created":"2022-01-27 01:13:21.000000000","uploader":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"ref":"refs/changes/25/826425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/25/826425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/25/826425/2"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-01-26 08:11:00.000000000","tz":480},"committer":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-01-27 01:12:43.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nremove unicode from code\n\nChange-Id: Ief9d24e83d74c1f7bf17c96abd713ee5b563361d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59858915222ce7b2670ea0226fcf8a50a2ea7fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59858915222ce7b2670ea0226fcf8a50a2ea7fac"}]},"branch":"refs/heads/master"},"b0950c64ee2e3a4eb2dbe0d6dc63f4421f82d2c0":{"kind":"REWORK","_number":3,"created":"2022-08-12 10:29:37.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/25/826425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/25/826425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/25/826425/3"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-01-26 08:11:00.000000000","tz":480},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-12 10:29:36.000000000","tz":0},"subject":"remove unicode from code","message":"remove unicode from code\n\nremove unicode from code\n\nChange-Id: Ief9d24e83d74c1f7bf17c96abd713ee5b563361d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b0950c64ee2e3a4eb2dbe0d6dc63f4421f82d2c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b0950c64ee2e3a4eb2dbe0d6dc63f4421f82d2c0"}]},"branch":"refs/heads/master"},"fbf80913644787e283b058eecd1ffd3aa9eb01b4":{"kind":"REWORK","_number":4,"created":"2022-08-12 14:41:59.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/25/826425/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/25/826425/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/826425/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/25/826425/4"}}},"commit":{"parents":[{"commit":"b2d6de569c7bba96cd3179189ffbcee6b7a28c1f","subject":"Fix unit test failures caused by type mismatch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b2d6de569c7bba96cd3179189ffbcee6b7a28c1f"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-01-26 08:11:00.000000000","tz":480},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-12 14:41:59.000000000","tz":0},"subject":"remove unicode from code","message":"remove unicode from code\n\nremove unicode from code\n\nChange-Id: Ief9d24e83d74c1f7bf17c96abd713ee5b563361d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fbf80913644787e283b058eecd1ffd3aa9eb01b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fbf80913644787e283b058eecd1ffd3aa9eb01b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
