)]}'
{"id":"openstack%2Fmistral~831960","triplet_id":"openstack%2Fmistral~master~Ie671f50e5ff5a7c746f9e95691eaf4dd19937b52","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2022-03-10 14:21:30.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-03-09 09:01:10.000000000","reason":"Adriano Petrich replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-09 13:00:43.000000000","reason":"Takashi Kajinami replied on the change"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-03-08 09:28:20.000000000","reason":"Eyal replied on the change"}},"hashtags":[],"change_id":"Ie671f50e5ff5a7c746f9e95691eaf4dd19937b52","subject":"Fix compatibility with oslo.context \u003e\u003d 4.0.0","status":"MERGED","created":"2022-03-04 15:07:11.000000000","updated":"2022-03-10 14:23:04.000000000","submitted":"2022-03-10 14:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831960","meta_rev_id":"f08cd8007003ad56d03e922e5d731d397bde13d8","_number":831960,"virtual_id_number":831960,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 14:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":2,"date":"2022-03-09 09:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-08 09:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":1,"date":"2022-03-09 09:01:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-04 15:33:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-03-04 15:33:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-03-04 15:33:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-03-04 19:29:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e640c06e78d291507feb2612e62359d7fc41653e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-04 15:07:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c5bda31d80c87ac7ee2735df8238aa55447a424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 19:29:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2502b1bece41de9457cb6a30180530 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/565c04cdb5e04800934e7ba5dbf046e8 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8edd9b973a714b82accff97430f982ab : SUCCESS in 12m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9d5e30b323c487eabb038bb0522bf50 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c52ce311a0407f8c15299c40b9ec4f : SUCCESS in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11c62e960af14916887aa09b6a109fff : SUCCESS in 10m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/44f4f84d01104aaba3ec5a67881657da : SUCCESS in 51m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42f0c83a429643e28e1eef239bd65a86 : SUCCESS in 51m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7099572caa5b41899f9a1ffd8cd0b8aa : SUCCESS in 48m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/be994866846d4ca396db8df18574cc11 : SUCCESS in 47m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2e4426856f2049a7a95c4b173a2baf40 : SUCCESS in 46m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9b1068274c245f1b931e0db07ae3048 : SUCCESS in 31m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4f8da62c9fcd427986a09f34cb699250 : SUCCESS in 28m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/45e9818ddd7d4b75b7aa07ce2daf2fc1 : TIMED_OUT in 2h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00c283015ff6abf1ce4bb81b1a992cb2d3d5231f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-08 09:28:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"353f4cb9acc91f579a2e2243440a046a84f03232","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-03-09 09:01:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e32e3d5fb652b387345711e2a7f6df4ad69e6314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 09:01:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2534e54e43cf7e12211feeaee0f120d5b0fd329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 09:58:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5ca1cf658f451699f55deaaf742dad : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92c6918c0bb0493780a044fdf70cf8b7 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee11f3bb3d1a4d84bda013fafc57c3ad : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163132dece54450189ac859da1c28fa1 : SUCCESS in 10m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/25278974b8554cdba3e9b4a3ca4f9fa4 : FAILURE in 40m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04ff1154a2694089856df5abb1935bdf : FAILURE in 41m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93396c09a0e7419cafc734e01d8e4afb : FAILURE in 51m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/110b8d26d20a4cd191f77391fe928b69 : FAILURE in 45m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bafae97f36dc446a8c81e3bd5e659cdb : FAILURE in 42m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ed14799b8e1e48e0a67a679eec17c601 : SUCCESS in 30m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/07ae98517f2041729cb11dede5ea2d15 : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7a1ef54e0c27926aa9e9534e2f85500e1efe8ba4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-09 13:00:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8aaa303cf8225c4ede2373554ffbe071c7f7e640","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-10 11:14:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d0f56dcd932ff6fcf32eee70074e1f38c8e016d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:26:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76db0c63b6154fe59cd45999e1161108 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2359a1a6d4f4db991993d2db04be552 : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e182eab4a074f0182ca57d3c010df8f : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b7f127626c14e7288f482cc26c2eecf : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/648f4c1c609a4d0fa4e94895d8c19b6c : SUCCESS in 14m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31cba9e40b814ed79745c65aaf3bbf4a : SUCCESS in 10m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b9bd51df111b45f798143077852dbbe4 : SUCCESS in 46m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2c7f05234ee4104b312deaad78a9f4f : SUCCESS in 47m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/525638f4dfe64cfcabcb1abbc401f5f4 : SUCCESS in 45m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0c77c3ee860a43758c17e08ebe4c0c3a : SUCCESS in 47m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/49ed507b3c2946049c33b9f5ecef0f44 : SUCCESS in 49m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a4a8f7306a1645988e6f995ca97b719d : SUCCESS in 30m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/becbbbb31f8545c7933ad4fa19e11032 : SUCCESS in 28m 35s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0a7b07fc0427440a967dcd232655b7e0 : TIMED_OUT in 2h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a43659582a7143c315d9e6d7ea93751b38950a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:27:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"782e0b6882f4771492e8edcd3dc84503880f295c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:21:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5a7e5dd858435fba78364a5b1c3498 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a00a5bce7d84dcfa58c86ce1feac4a9 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a803c7818a744bc861c1df5f59be7f6 : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3352f398f3974a22bd1856255235d2b3 : SUCCESS in 10m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/adbbe2e388fe4da0ade4f67c72a350b0 : SUCCESS in 50m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/184951ca3f6d41d28bfd723e56d0558e : SUCCESS in 41m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82f43385df4741beaa575782fe4286eb : SUCCESS in 46m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/05c66ea60d2043c292532d916853a2cf : SUCCESS in 41m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/59f8a487016d4f898eb6eaede6c66a34 : SUCCESS in 46m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8a9344c07141483d9d3e42c6c2f74b40 : SUCCESS in 31m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ee983ed0212b4871ac92f4fb1fa1a5c7 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9e327f5c1dcfd6e08224dd9f2de89d2b31d263e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:21:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f08cd8007003ad56d03e922e5d731d397bde13d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:23:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda588470a7d4801ab339af188cc8340 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e07dcc424f87d3afbfa90d781480a197f4dec1d","revisions":{"3e07dcc424f87d3afbfa90d781480a197f4dec1d":{"kind":"REWORK","_number":1,"created":"2022-03-04 15:07:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/831960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/831960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/831960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/831960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/831960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/831960/1"}}},"commit":{"parents":[{"commit":"e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b","subject":"Replace deprecated arguments of RequestContext","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e82c4ca08b7195b3e907bcf5192a45d3e5a1e31b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-04 15:05:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-04 15:05:06.000000000","tz":540},"subject":"Fix compatibility with oslo.context \u003e\u003d 4.0.0","message":"Fix compatibility with oslo.context \u003e\u003d 4.0.0\n\nThe tenant argument of RequestContext is no longer available since\noslo.context \u003e\u003d 4.0.0 . This change fixes the compatibility issue\ncaused by that removal.\n\nNote that this still keeps reference to \u0027tenant\u0027 argument to make\nthe code compatible with older oslo.context, but that can be removed\nonce oslo.context \u003e\u003d 4.0.0 becomes available in upper-constraints.\n\nChange-Id: Ie671f50e5ff5a7c746f9e95691eaf4dd19937b52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3e07dcc424f87d3afbfa90d781480a197f4dec1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3e07dcc424f87d3afbfa90d781480a197f4dec1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
