)]}'
{"id":"openstack%2Fmistral~844679","triplet_id":"openstack%2Fmistral~master~I4e27be997dc35592813d89a211bce553cd7252d3","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"27190":{"account":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"last_update":"2022-08-14 19:46:58.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-14 17:02:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4e27be997dc35592813d89a211bce553cd7252d3","subject":"Update python testing as per zed cycle testing runtime","status":"MERGED","created":"2022-06-04 07:20:26.000000000","updated":"2022-08-14 19:48:32.000000000","submitted":"2022-08-14 19:46:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844679","meta_rev_id":"2399d619d679fb66c1b1b7afb3f1049976227869","_number":844679,"virtual_id_number":844679,"owner":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-14 19:46:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"recommended":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"all":[{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-08-13 15:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-12 14:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-12 14:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-04 09:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 05:54:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-08-12 14:37:41.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-08-13 15:07:26.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"b8b2d3f3d539629f57429f6e1566dd2334a72aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"date":"2022-06-04 07:20:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81cf00442131371ad82dd903eb796e68254bfcd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 09:36:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea3b306aefa6461ca8eb1da0d536439d : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6c27d45d2314bbdad932ba9dff0833e : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dece56af3ce0444a87d2bc8c69ddbfed : FAILURE in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309154f394ed40df8cfb73187c6e1069 : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/157f3fa0543e45f8af001528c08517ee : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67dd178a4dc420dbe246197e38175a0 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a1d2f55b0b841fcb57e21a4320ff5cb : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc1f5750ca8941c7b1e65045532f39c7 : SUCCESS in 9m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e3c78dcc595348be9a10cafa9f0f36de : FAILURE in 39m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcd42bf7f4064f82a055875b0bc5a204 : FAILURE in 37m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e742c83209ab416fbe25b39ba3aab253 : SUCCESS in 48m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1525813eff884c4585185ce0d96fe6fa : SUCCESS in 48m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7ddd827db47a424aab67fe2099a87e4c : FAILURE in 37m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aecfd87df4fa4192ac06db21dd4f01ff : SUCCESS in 28m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38b3eef0c3e542ec8a4a3703195604bf : SUCCESS in 28m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1881247f2fbb4af8ba54ce01e8db16d4 : TIMED_OUT in 2h 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae3cae85043add6934b6a61fa37df883d6be05b6","author":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"date":"2022-06-05 02:05:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b1fdeb6faea4db113c567a1875075b90e883d52","author":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"date":"2022-06-05 02:06:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d3a61d248998f9907014194ebaeba212cc8442de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-05 05:13:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c0660a206a5942239593b6bd9e9dfd35 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e723db62fc442bb06bdb86f8a4df6c : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a03635c0374f45bcbbf54ee0d7c92191 : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c268b703b2b94bf9b3e6af328d3f14d1 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afff89c0e27d45c193d151596ab3a5ba : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c1a6e9582441e8ad38ae8ff0a2dba8 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af0efd624ff844569c7a3843bcd08c8d : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18726af0c37d41879b1fccc9dc4c9b16 : SUCCESS in 10m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a53d4f55e2634da08e8dc3c03c55362a : SUCCESS in 40m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e06ce789a914f51b6b4714598dcd5d2 : FAILURE in 40m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96fa1316efb94d4fa58e00e18a9649a5 : SUCCESS in 44m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c2567eb3c27b421cb490a9b9fa2453a3 : SUCCESS in 40m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/60b8e622f28f4fa5a5751e809f7bcb14 : FAILURE in 40m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5aeb3a38e868403eb42c0b5d36bd18c9 : SUCCESS in 28m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8cfbbd20d0a64c6f9d253173ce6c7182 : SUCCESS in 26m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e26a0665d5eb499f8d751d16b83bbb54 : TIMED_OUT in 2h 50m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e9fa90ce7d970ee7ebe547037646f91a11dcdb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-02 05:53:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4d43e3173004477a2aeaec23a8e812acdeb5d5d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-02 05:54:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de7075b4df71210405c739030b00d5fc909ffdc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-02 05:56:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1949550d90f838312b75d43c6b890e3096db74f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 08:05:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c0686b08bdb451ca7d26a41f32dc3b6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b23a003073a4006a94692679d0eb162 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a583311e823e4d38ad5362d7d428267c : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6347c4882cba4bc3a1c6dcf581d9c145 : FAILURE in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fee0748a7dae49309e22fde325ff4563 : FAILURE in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c341f230607c42fd8404150c93181011 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3275d18e5d541138a11d4b8a869402d : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e891f423523045f5b25025933da9a903 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e97156c0b9f41e096b3bdadca51939e : FAILURE in 4m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ee414711046f492e822a66a0a1db5bf1 : SUCCESS in 37m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31687edccd4e4d12b26257e09e7f6e49 : SUCCESS in 43m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50e9c72068af4f4fb51aac464734e349 : SUCCESS in 48m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f33236f89e6a4e698977d1f4e0d1e435 : SUCCESS in 48m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/746321b42f9d4735a04d68aa4ee4f542 : SUCCESS in 45m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/11b2ea3669f4497384c095bc225c7b6e : FAILURE in 32m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e4a29f97592c439786611294fa4fca0a : FAILURE in 33m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9efc41a40bc44efb8806918c85d608d0 : TIMED_OUT in 2h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2a4e2e0564e78b0c52ae288631037ba87981fdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-12 11:57:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3dcd2bc54a5c4d92ad58e910737a56558995db81","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-12 11:58:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"378b529489692fe07467884a9eea4ba66a466405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 14:18:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebb364b36eed48969ff12d606ec15488\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef0b185a7f054b7f8953dcddec1cf2f1 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d3fb373f724c9e9cd36cba3a380147 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91ad75998c184e9c80b647f3c2f3e6e2 : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcecb9573888415f8be61d6670c218ae : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b59c501067f040bf81f22e5addd756e8 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7096e4dfe64dd48a4c37976b236ebd : SUCCESS in 10m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fafab5a0b7a940e2bd07adb3d89cd436 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/781da97366ef4702be69d4e44fe5dffc : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5436d27b9bd04aa0821683ce9cebbfd2 : SUCCESS in 13m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2bccbd68de834aac9200c63fd3ca1290 : SUCCESS in 53m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d7b72633374a3a9cb1581802a99ba0 : SUCCESS in 52m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74e689fdaa8c4ab2b27142f842398654 : SUCCESS in 41m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/87999a0899194712b31c0018d28c4e5c : SUCCESS in 51m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/792a6be8d6774bc79a4d64bba1047fe4 : SUCCESS in 46m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/564bf42d496e47fea4a5d635374ca13e : SUCCESS in 34m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3d5be979424648509d1e1012fbcbb2b3 : SUCCESS in 30m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/99a6f0908cda4a20953e91d6ea767393 : TIMED_OUT in 2h 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b885cdba9fb2523e945c04887cf8002bf3fe8c0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 14:37:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ecbab3614c02db65198a49a6cc7df7a5f7d84e40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 14:38:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"75a809012e6b0cd14c4d59583ea23dd0e9737a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 15:21:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d177e4c6b684b83ac2a8ad1e3e60d07\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21c40b7f97ab42d3887c3d02494b46fc : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3a9d7e48f94a5f95d2e8e302ac5f77 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad2b70766c2343d29e8587b51154e20b : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7337d0de065442a4bb772ab87b0530d7 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52254555707c4490abd42170d1407d51 : SUCCESS in 8m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d30afc464f04c0b8eeea980fff5f465 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faf51b57742a4a47b6ff10e9168c065a : SUCCESS in 5m 03s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4ae96ba0849c4fd48fe279cda8ddd1fd : SUCCESS in 38m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3501f110d004107a95db383d6a12ddc : SUCCESS in 38m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecd75c9f16544aa18184a0f2e1a8620e : SUCCESS in 38m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/63c762d557f0491e916be1eab0a4859f : SUCCESS in 39m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/424b056b9a3349a68081036fe9989867 : SUCCESS in 41m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9b6df9aa31a4493a93ef809fb1c73950 : FAILURE in 33m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e5f152d518ea4ad59ea61adad350d472 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":4},{"id":"48e66e2102ce2707e61213b373a23e2ea75d997e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 17:14:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e5a2508d2bdcd22ad4d96111ed1eed5f0199dc1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 19:24:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41f16a211fa34388a524f4510bef1cac\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a43b0f79016d4363a1a49fee6435fc3d : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78fc830efb7c4989829b4f9574a383b9 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9249eb01578b491abdc9f2aaa89f56ac : SUCCESS in 9m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fa9313a139e424da07ec1ffda33bb68 : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d078ec81f24347d38c34285445a34ac6 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28198d776414bb58ba4735f995593e1 : SUCCESS in 7m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2b8ee346ad245d9aa4227e8d86b7200 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286df2e789814505b3da8204216757a6 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aefd299cd425463d8d481009ecd42e92 : SUCCESS in 10m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/427e7222b1554808b1dbf5faa838eeff : SUCCESS in 37m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/286cccffe1614575ad0c3c79158cc5fc : SUCCESS in 40m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/313903c1d7ee4895b4d2a6485e8fc808 : SUCCESS in 39m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b28ae5746f024ab993dd404f1cf68eac : SUCCESS in 39m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0acc2f9752cc494c814285bdc3badf5e : SUCCESS in 39m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ebb16f7051c5455eb7a3816c20a2178d : FAILURE in 33m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ed6ff1a4522f40ddbfe4714b102c2d24 : SUCCESS in 33m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2d598460387c470d906400727d32c58f : TIMED_OUT in 2h 06m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4bc26aa0019082ff370eb61480a1a51599b6f2b6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-13 10:36:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"abe9a6f3656a8e2cbc1cf0ae9f6aa38914f664fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 12:37:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d2df586d4a4cc392523f4bfdeeb3f5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/93465d30f3ce42deb4f96597d5f857e9 : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84e2afc014f4ac79879ff490b1b06ef : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/122ffb27e7fb40e3848a47e73e5006e3 : SUCCESS in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c124f5aaaa1644a29961d3277985d48a : SUCCESS in 8m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/519c33335b004a30a9ed09fc79f57577 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b739e469a33f4e68a1f279d0b6c0dedf : SUCCESS in 6m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/76905ad7b3e54ad680d5737ee402ca0e : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ebaf31518264de3a86c2915b91ca2c4 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09318a1ea053412eb0e504b3bc5405ac : SUCCESS in 10m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c69d6d13ef95405f9daaa950a8aedfe9 : SUCCESS in 43m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/118aaa5dfe7a4225aca76be9e0a27e57 : SUCCESS in 42m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33431f03f64d44c0bb32d679927da60e : SUCCESS in 52m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c99effc2aacf49daa1ecc8b522a2ac92 : SUCCESS in 44m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d2b3d8d1dca145caa431d38ff49cd78f : SUCCESS in 43m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/37db90a9bbf3478b8fcfe9c1f9e07b9a : SUCCESS in 35m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4ef3772aa2e748cc8a9dc548ab7a3935 : SUCCESS in 32m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0c0170f4310c43d794a7f1adbcb8cbc1 : FAILURE in 1h 54m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a2ccd366ceff56806818508079dddd752299f570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 12:38:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e73aa18236b4244ad07bd707652329ff4fb54a55","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-13 13:26:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"8b143179e31a063745e0f04ff62f243e799b60cc","tag":"autogenerated:gerrit:restore","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-13 13:26:37.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"234b98c83640fd36698cc71ea154e6df478bf553","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-08-13 15:07:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"77b819b44b82d2b031c8628baee8869f311fdc72","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-13 15:14:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"19abac078b8bf4d84ec986d94e7f35f5a42a9162","tag":"autogenerated:gerrit:restore","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-13 15:14:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"88885561f1e54cd2b2d9422837dc476658227975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 17:13:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/387cafc0a60a4f0ebb31fda9c41cf3cb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a47a457312c8457d9956e4336fb6e7a7 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993ae1a1c33c4b2da1a8c47f7a67572e : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec6f2352ca3c49a8bf9bc1b5d2dce989 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a174ad54d0845d28f387480c31a4608 : SUCCESS in 8m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ec6bea993ed40299e8fbe32ccd7e446 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2972ec653746a68ef3f7b1cb8d829d : SUCCESS in 6m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a62045465e8e467fa18d445c6cd2969b : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc69f418db1441b78b908f9a41326f67 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/171acbc1be8249c4824f719fa376da1d : SUCCESS in 10m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d245a1825e7c464190d61cf081615170 : SUCCESS in 33m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/728a6535a7f143ee972673ca04a4727d : SUCCESS in 36m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cda8ed0c938486b84d85a8aa3ca068e : SUCCESS in 33m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7499c7141dac4b659f65f52f7775e751 : SUCCESS in 32m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3416fde421a14c969ee71cbf4cb033ee : SUCCESS in 36m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e0ea91350f3744e4b311aaaf12195d2f : FAILURE in 31m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f5b6509a70554e289f1cf809c4a0d1ac : SUCCESS in 34m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a24e9b607fb487887fa030626f24bfb : SUCCESS in 1h 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"196be7a27c15e3112c2c43658845aa845b87242e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-14 17:02:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f74a61ecf2670d34d7de5cd8f7d0670bda31d8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 19:07:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/302395cc884940559637ebbd8566da70\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/60cdb06b02b946dfba424261bcb97b01 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ce7a6196e648f98912da34f7596eaa : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94e1101547224895b94de476810ae6d6 : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eda649c1390145ddb0afce6a93580927 : SUCCESS in 9m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9296ea8126f24dfb80191d535606aae6 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75421b3a6e6465694464759eb0f6069 : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24c9ceef3cbc44b2b7cea0bc0fda9b20 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11896b4d6c7c427a92077924e724a1b3 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc2c17b0326f4b04a9483967ad280bb2 : SUCCESS in 13m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/acdd85bcb4f0449c899961d230a7d820 : SUCCESS in 56m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3ea769adb0640faa4f78f2aaab2488a : SUCCESS in 35m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b57065a23e34f268dd2c26f03795962 : SUCCESS in 57m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/117f7ccb78ca49cdaaa9b145127668ca : SUCCESS in 33m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7004e5ca3fee4d3dbb0dd854eb4dc256 : SUCCESS in 57m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c7f38c13055b477cb72e873de9f516d5 : SUCCESS in 31m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c4b0b982a348475396630abfa3af04f2 : SUCCESS in 33m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e68bf3dcce844a688a5367d56a3c90ae : SUCCESS in 2h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b29dfd7684468585dc1eed5cd6fbc7e39759db3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 19:07:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c3caf38871f9b8033e095178f47f47410f397315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 19:46:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cabccc5ec22d4e899e1927040ed87f91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a40545b61f441ac8ba07692e574ccbd : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9703a46733974a5fa481a1024c318e52 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc64c40d933412cbaf6542c3dcb873f : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b32f36e1a7394289aac497290334c75e : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e11ef5683394a2bb51c5313478affc2 : SUCCESS in 8m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f546e79b02c5499d8c5047f498c18413 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/886003c9ea034b88b011f9ed9c1ea423 : SUCCESS in 4m 43s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/442bc97ef9024683b6d4ccb7e875b5bc : SUCCESS in 38m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ecbc3f96920488ab350a8e741751589 : SUCCESS in 37m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f42c39b09e8841e4a1ed464dadf0f804 : SUCCESS in 33m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/399672d3d96545189f69ad6863c8c786 : SUCCESS in 35m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c561d49b36074a3a9d600a80918eaa1f : SUCCESS in 34m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/99e1033c2e66460d9261258c2f4da3f8 : SUCCESS in 31m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b97f83dc19d049aabdec30fdaadf95fb : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ae36564e1f054b0a7ddd95b581537aae0263dd03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 19:46:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2399d619d679fb66c1b1b7afb3f1049976227869","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 19:48:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ca568735c614023b9b62d10749d59e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaca99bf5e0b475e8ed4a50150544c62 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44d8999e460447dc87452223ee45cd61 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"88c7a5d1ff7c0abafa1fd776dff0ce14a4241ca9","revisions":{"1a441ab78a4a49485ef262a2650c9f09d3833d51":{"kind":"REWORK","_number":1,"created":"2022-06-04 07:20:26.000000000","uploader":{"_account_id":27190,"name":"Leopard Ma","email":"mabao@inspur.com","username":"LeopardMa"},"ref":"refs/changes/79/844679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/79/844679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/79/844679/1"}}},"commit":{"parents":[{"commit":"28f94b0c64ebbc04729c441c801682e312fc9c64","subject":"Merge \"Fix compatibility with oslo.context \u003e\u003d 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28f94b0c64ebbc04729c441c801682e312fc9c64"}]}],"author":{"name":"LeopardMa","email":"mabao@inspur.com","date":"2022-06-04 07:16:04.000000000","tz":480},"committer":{"name":"LeopardMa","email":"mabao@inspur.com","date":"2022-06-04 07:16:04.000000000","tz":480},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.htm\n\nChange-Id: I4e27be997dc35592813d89a211bce553cd7252d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1a441ab78a4a49485ef262a2650c9f09d3833d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1a441ab78a4a49485ef262a2650c9f09d3833d51"}]},"branch":"refs/heads/master"},"559909571150061de41a441692277cb8f7f07892":{"kind":"REWORK","_number":2,"created":"2022-08-02 05:53:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/844679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/79/844679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/79/844679/2"}}},"commit":{"parents":[{"commit":"28f94b0c64ebbc04729c441c801682e312fc9c64","subject":"Merge \"Fix compatibility with oslo.context \u003e\u003d 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28f94b0c64ebbc04729c441c801682e312fc9c64"}]}],"author":{"name":"LeopardMa","email":"mabao@inspur.com","date":"2022-06-04 07:16:04.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-02 05:52:51.000000000","tz":540},"subject":"Update python testing as per zed cycle teting runtime","message":"Update python testing as per zed cycle teting runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Removing the py36 centos8 job as well as\nupdating the python classifier also to reflect the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.htm\n\nChange-Id: I4e27be997dc35592813d89a211bce553cd7252d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/559909571150061de41a441692277cb8f7f07892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/559909571150061de41a441692277cb8f7f07892"}]},"branch":"refs/heads/master"},"420d2f4352933cb2dbe10dd37120feaf41c31607":{"kind":"REWORK","_number":3,"created":"2022-08-02 05:56:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/844679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/79/844679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/79/844679/3"}}},"commit":{"parents":[{"commit":"28f94b0c64ebbc04729c441c801682e312fc9c64","subject":"Merge \"Fix compatibility with oslo.context \u003e\u003d 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28f94b0c64ebbc04729c441c801682e312fc9c64"}]}],"author":{"name":"LeopardMa","email":"mabao@inspur.com","date":"2022-06-04 07:16:04.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-02 05:55:56.000000000","tz":540},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Add release notes and update the python\nclassifier for the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.htm\n\nChange-Id: I4e27be997dc35592813d89a211bce553cd7252d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/420d2f4352933cb2dbe10dd37120feaf41c31607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/420d2f4352933cb2dbe10dd37120feaf41c31607"}]},"branch":"refs/heads/master"},"88c7a5d1ff7c0abafa1fd776dff0ce14a4241ca9":{"kind":"REWORK","_number":4,"created":"2022-08-12 11:57:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/844679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/79/844679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/79/844679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/79/844679/4"}}},"commit":{"parents":[{"commit":"28f94b0c64ebbc04729c441c801682e312fc9c64","subject":"Merge \"Fix compatibility with oslo.context \u003e\u003d 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28f94b0c64ebbc04729c441c801682e312fc9c64"}]}],"author":{"name":"LeopardMa","email":"mabao@inspur.com","date":"2022-06-04 07:16:04.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-12 11:57:06.000000000","tz":540},"subject":"Update python testing as per zed cycle testing runtime","message":"Update python testing as per zed cycle testing runtime\n\nIn Zed cycle, we have dropped the python 3.6/3.7[1] testing\nand its support. Add release notes and update the python\nclassifier for the same.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.htm\n\nChange-Id: I4e27be997dc35592813d89a211bce553cd7252d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/88c7a5d1ff7c0abafa1fd776dff0ce14a4241ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/88c7a5d1ff7c0abafa1fd776dff0ce14a4241ca9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
