)]}'
{"id":"openstack%2Fmistral~852951","triplet_id":"openstack%2Fmistral~stable%2Fyoga~I11caebaf5979ae4774139ddc8f7f98d64ebaffeb","project":"openstack/mistral","branch":"stable/yoga","topic":"add-yoga-python-jobtemplates","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-12 16:10:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11caebaf5979ae4774139ddc8f7f98d64ebaffeb","subject":"Add Python3 yoga unit tests","status":"MERGED","created":"2022-08-12 11:59:53.000000000","updated":"2022-08-12 16:12:20.000000000","submitted":"2022-08-12 16:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852951-add-yoga-python-jobtemplates","cherry_pick_of_change":809821,"cherry_pick_of_patch_set":1,"meta_rev_id":"537d9f226fca3a8e8aa178b0a51b9af382854fa6","_number":852951,"virtual_id_number":852951,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 16:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-12 14:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-12 14:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-12 11:59:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-12 14:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 14:37:54.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"8b760dfb765a0ecf3ffc065fba19b522d6084b07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-12 11:59:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b519f053d4aabce582ee6bd55821edaf8527b12","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-12 12:00:05.000000000","message":"Topic add-yoga-python-jobtemplates-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"6592b0b43b7e8cff30521194b562f9c695a23c85","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-12 12:00:07.000000000","message":"Topic set to add-yoga-python-jobtemplates","accounts_in_message":[],"_revision_number":1},{"id":"c1f94868ffc90892decd38000fcf6b09706ed74d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 14:11:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78bec1627a240fb9ef6a12ebca8f214\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b918fcc417ab420d864d1bc83ba8a7c1 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69ffeb3a6bec408394661c4f1ed5ff5e : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b7372c4e1684586b34909c4ade84650 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bee9b243e0c47a79503d97008db5a12 : SUCCESS in 8m 08s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/485d3f56dfb14021b20486763dffa3d8 : SUCCESS in 54m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4274f714653c4d73b88ea182b35278d6 : SUCCESS in 1h 00m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a060e101dd7a490090e08b9aec34c3d8 : SUCCESS in 51m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1854c5724ce944ec8b0040d7caa1832a : SUCCESS in 59m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e18144da8ac4d9093e41fa7700e6509 : SUCCESS in 54m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1fa718c7a0ce496ba36b738ce59cda3b : SUCCESS in 29m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/963491ddbd5a4239b0d657d320d3a501 : SUCCESS in 26m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/050ec62c5a9c413b9965bca06be4e82f : TIMED_OUT in 2h 05m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d72d3f4463ced482da9d85200a34a080a3212bc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-12 14:37:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a69d9e96a95cb1173a6667bf12a1ea0924892eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 14:38:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"753c8e96494ca0128afd477ba4bedaf4d63219ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 16:10:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28434e5805cb4848bd215e843f8f7ade\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838a02d66d7b475cad315b8f1d0a4b7b : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98ac18297de146b4ae72f19b40b248eb : SUCCESS in 8m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec880124dfe54531844823b96401fc2c : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fad9f271fe34025878323fe4bd15561 : SUCCESS in 7m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/866adc060ffc4df88f8afdec50eabb10 : SUCCESS in 53m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b209284324be4a07b7634c26b1f9d4e2 : SUCCESS in 52m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98fa48064ea74371b28546c5d772f226 : SUCCESS in 49m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/511426956ca1431aada7272b4d17df76 : SUCCESS in 51m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3f5f06579a07439385aa4eab3dcf2c3f : SUCCESS in 51m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bcdeaa18aaeb4ed99eb34b6dc37f9552 : SUCCESS in 29m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/01b9b1e23b8b49918967a450449971e8 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8ff5d6665a022743518c13172124333fc55c3acf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 16:10:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"537d9f226fca3a8e8aa178b0a51b9af382854fa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 16:12:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0533f29389774fc784b1bbb593f02e73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffb631d4f154035b09397f39a5c9020 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21867b48c7f223c39fb00b244df49898d337dee2","revisions":{"21867b48c7f223c39fb00b244df49898d337dee2":{"kind":"REWORK","_number":1,"created":"2022-08-12 11:59:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/852951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/51/852951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/852951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/852951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/852951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/51/852951/1"}}},"commit":{"parents":[{"commit":"a9024d42ff7f147e1965d42852fd1f0bddef7ca6","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9024d42ff7f147e1965d42852fd1f0bddef7ca6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:22:43.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-12 11:59:53.000000000","tz":0},"subject":"Add Python3 yoga unit tests","message":"Add Python3 yoga unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for yoga.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I11caebaf5979ae4774139ddc8f7f98d64ebaffeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21867b48c7f223c39fb00b244df49898d337dee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/21867b48c7f223c39fb00b244df49898d337dee2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
