)]}'
{"id":"openstack%2Fmistral~854240","triplet_id":"openstack%2Fmistral~master~I1bc20b953787cae9b42d12bb5eb3c7a94ed4bbdd","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2022-09-09 13:10:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1bc20b953787cae9b42d12bb5eb3c7a94ed4bbdd","subject":"Do not guess content from extension on API","status":"MERGED","created":"2022-08-23 16:15:31.000000000","updated":"2022-09-09 13:11:11.000000000","submitted":"2022-09-09 13:10:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854240","meta_rev_id":"46590df8f83603b92faae29795abd9324f438831","_number":854240,"virtual_id_number":854240,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 13:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-09 10:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-09 10:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 16:16:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-08-23 16:16:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-08-23 16:16:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2022-08-23 16:16:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-08-23 16:16:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-08-23 18:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9c81978247e6a3f9c46ef028417630336308ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-23 16:15:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24f4b6cd2c79f4227ea584ed2356d660d261c9e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 18:23:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/feec47ca2385439aa01ae9c21e88ddbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bdde111b3d740ea9136504454c5137f : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d5b86f00f664136b51f34ac6167e4f0 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/671eb7318f5748cba8ce34f4ec78aafb : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea0a8ba82f1844e4a004014c2ac25673 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb9fd091faf43e4b183199ceb640a44 : SUCCESS in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e0f7f1aee8049288745b5b5d86ec8a2 : SUCCESS in 11m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8ded46abfdbb416f93b3845105b4076c : SUCCESS in 49m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63b58002976e462da3a6825d9d7daa76 : SUCCESS in 40m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7f8157db0b943a8921fdf9a46d99216 : SUCCESS in 42m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b40341a1d1e148d29974d95334b4faa2 : SUCCESS in 38m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a2839fc0fd4342ac81a4ad83018a9e34 : SUCCESS in 46m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1beb08bed28c4976b71e02a286877697 : FAILURE in 35m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/80505b3ec9874f878c08c92a645fffb6 : SUCCESS in 32m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/64ece34ec16d483da3a011bc486e9b9d : SUCCESS in 2h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6fa2e565eaa688bb21bbe6d404768a3852af2f1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-24 09:07:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6659dd8f5590fc061fb54850d364aa8f77f44cd9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-29 14:20:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8363037d72ab077caf0dbc58ac65156b6b31890b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-29 15:02:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae2602eb99b1557e3e810cab791be72bae5e2d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:29:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d23ffdffc2d142c695ed530fda8d2bd7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f199081eb2741bf8aff8494985851c3 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36449f84c0e74beeb949ebd41dda08ec : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4753e9feeb974abd821647f620a42a05 : SUCCESS in 10m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/279f0c72f9d544c08be3f4991d3e76d5 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0331f49905374224a6069aead77a11ed : SUCCESS in 10m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44c5ecd3cf494f6d86b8218d6943af39 : SUCCESS in 10m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4beaf77a07d94c4085330a4ca2f18708 : SUCCESS in 45m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11e25517e9254af8ad6bdbe95f0271fc : SUCCESS in 37m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f38f093616b642fd833dfd4ef29132e2 : SUCCESS in 45m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2b3473673471485faffcacf9966875cc : SUCCESS in 49m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c0bad3a88b7c4c5199e46c4cd019e98b : SUCCESS in 37m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5b5af023bb12424d854fb7f218c3aef9 : SUCCESS in 35m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/398d09ae9e0b41fdb5f3b25661efce8e : FAILURE in 29m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/84c5be158ce04ee2a804509d205e5e43 : TIMED_OUT in 2h 06m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d6ee61b502cd428b2157a5f249695785fc9c31c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-30 08:45:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f4795c968d6c5613c56597b05d214170f53dc7db","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-30 08:46:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62f6106c56266c02f4f83e2e806ec5a2ef29b1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 10:43:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47d17261d8d74097aef84ea9b9f9dbc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b59f5e66b4b4f1fa45aeac8f9a8f2bc : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b234e1bea0b4f6697c64e15145ce2da : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4c6316c91da45cfb25cd38c7eab00b3 : SUCCESS in 9m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6332df6e46c4a5dbe8d80bc3a2fd4de : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0695f2a5e434428693e70f50cc46b785 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/348e2aae880d45deb18341f6b927a6ea : SUCCESS in 11m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/325660dff6684ed293cc0d60fdbe30ef : SUCCESS in 51m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/042fcc3683774ddbb1ed7c10bc1b9a6c : SUCCESS in 51m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c873117eed4462e9ca4ff52c223a9ec : SUCCESS in 52m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a03e60645db94a86a7074fe757f6dcf8 : SUCCESS in 52m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/117892b2123549de9ce304124ac4fc6a : SUCCESS in 53m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/103003a856ad4db09a48fafd7922b4d6 : SUCCESS in 35m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3387ed20669c4481b84578fd00f296ad : SUCCESS in 29m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e845abfc5709447ea52e3b26c6eee433 : FAILURE in 1h 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0311d60ede9d8c2b024792550d786e1cc9db4efa","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-09-02 14:06:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0857fc0566ef5f5189a03d3770a00ce11ebbb333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:08:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6a8a6dadf0ea4a38b71ac0389ba4574882b2362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:57:51.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"c0b2ae3d77d842c632645ce10d490dbf1dd50e5c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-09-02 15:17:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f4dbe3d449e7b898bb09f5ec6335c3a48ff7ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 17:04:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9b016fb240d4fa98bff8f415a8d01df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/054ce8e7e15e4f36af24c0aee0ceb90f : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ff4f293acd549b391041ce1bd4ca074 : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07b218ca1eb14971bacde0a26576a647 : SUCCESS in 11m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca4be74c2bca43b7b396606298d25410 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92936cc962b04a8b83568dac35c7869c : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/336bd1de856b494ab384d5e4eccd746b : SUCCESS in 10m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/213e1e5de02c4c07b22c0f25f76029f2 : SUCCESS in 44m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27998044457b41b3b6718bc36a8621cf : SUCCESS in 49m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12318bfd641a4e029d029a90575a7e4c : SUCCESS in 44m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/93087d1645ea4587b8980deec596e886 : SUCCESS in 41m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3b385610767344c5ba4714f848d4d38f : SUCCESS in 45m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f1e93431bfce44dda96044eba8a52351 : FAILURE in 35m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/667ab57d0a8f43dcabcdd77738fe3b19 : SUCCESS in 34m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c94411f4c9bd4a168f0fd621bb030a58 : FAILURE in 1h 45m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04afb0dfe266a0726a5ad71ebed5dd88bfe6c00b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-09-09 08:41:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"db35473b215f22a7299e7cfd7c78f4aa1fa90a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 10:54:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf818295c3a4a118a2e9f94cc4b145e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d43867fec0845f2ba8c43c755ca47c6 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c151761e58c5466f8abe57c2b699466e : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1aa4855e518d42e1a94200b8b48ccc46 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68cd5d8fa13844cbb41ed7aef5581875 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d87191502154700a8ebc59b5ce6172a : SUCCESS in 8m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/426fcb03b7eb403d800e9f961be68763 : SUCCESS in 13m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6645cf822b1e4299a6beee6e77bfa64b : SUCCESS in 43m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fc20b0b601d440b93bbc9147720ccf8 : SUCCESS in 43m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14e439810bd74012b0ee9740e1b4d3d7 : SUCCESS in 51m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a33034ef82e94f2ea1b7c037f712b831 : SUCCESS in 42m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9e726d73fb704dd6bd5e818cec6ef8c3 : SUCCESS in 49m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d99ec130e82e4041836c639a5c39fe4e : SUCCESS in 40m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f39070dcd4d04927a47339a6779e60e1 : SUCCESS in 34m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fa7058088cec4afa8348570afbe089d5 : TIMED_OUT in 2h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65cb88dd26b46abfbc98aa52faa60f2319498bdb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-09-09 10:55:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"329ec2423bdcee0c38531a643178153cc759bb9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 12:06:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a770139d07c2e947c8d21bd3bdefcb864c03d274","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 13:10:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1939f8d622854f37a1312be5b6ffe357\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27087bb96294e6bb41357a5058a77a9 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a06f4ddf95f4c0198d7e9456fa5213b : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72dfaa7b44bd41b8bd4035a993f71f12 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4647d839f38e46a1b3c31c8128c05e3f : SUCCESS in 7m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7bd8bf70893445f799bee5bb521b397f : SUCCESS in 41m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4db28dc7f46410e807dd652ea3ea1c9 : SUCCESS in 56m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e7f051a29434ed187a487301b721453 : SUCCESS in 56m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5130b5ac2e29468798da6127c619f6e7 : SUCCESS in 51m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9df66e15fcc04a0886eb6b913e212472 : SUCCESS in 57m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/00b4da215a8f49abbdb9ee27a28bdd2a : SUCCESS in 40m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6f97093cecb04fef94faf7b57d02fba6 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":3},{"id":"53a6e4d850d5f9d2453709f2ba7f8665d811b7d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 13:10:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"46590df8f83603b92faae29795abd9324f438831","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 13:11:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1cd1fa600364f4b9f19542baadc9fe6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3dbf05ad46f46d4b0249fb20edbb3c5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2a312cc81446c9e3ac98e6bc698fb44a82cba2b0","revisions":{"f82bdf1933caf51d74b406cd7654c8ab8bff63f0":{"kind":"REWORK","_number":1,"created":"2022-08-23 16:15:31.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/40/854240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/854240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/854240/1"}}},"commit":{"parents":[{"commit":"c7d5bc8b083341ad8b9147ca294950f3f72fb973","subject":"Merge \"Replace deprecated assertDictContainsSubset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c7d5bc8b083341ad8b9147ca294950f3f72fb973"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-23 16:11:39.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-23 16:15:18.000000000","tz":120},"subject":"Do not guess content from extension on API","message":"Do not guess content from extension on API\n\nWhen getting a cron trigger (but also any other object) by name, pecan\ncore will try to guess the content_type from the extension.\nWhile for most name, the guess will fail, it may succeed, e.g. when a\nname is ending with \u0027.pl\u0027.\nPecan guess it as a perl file and then reject the call (with a 404\nerror).\n\nThe fix is to tell pecan to not guess any content_type from extension by\nsetting guess_content_type_from_ext to False on initialisation.\n\nCloses-Bug: 1987418\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I1bc20b953787cae9b42d12bb5eb3c7a94ed4bbdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f82bdf1933caf51d74b406cd7654c8ab8bff63f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f82bdf1933caf51d74b406cd7654c8ab8bff63f0"}]},"branch":"refs/heads/master"},"2ffbb00537104e52541ac7979bbb06c2e65ba682":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-30 08:45:04.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/40/854240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/854240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/854240/2"}}},"commit":{"parents":[{"commit":"60c9dbf078df309a14da8c43d65f3afbe3dcaeb3","subject":"Fix race condition when cancelling workflows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60c9dbf078df309a14da8c43d65f3afbe3dcaeb3"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-23 16:11:39.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-30 08:44:49.000000000","tz":120},"subject":"Do not guess content from extension on API","message":"Do not guess content from extension on API\n\nWhen getting a cron trigger (but also any other object) by name, pecan\ncore will try to guess the content_type from the extension.\nWhile for most name, the guess will fail, it may succeed, e.g. when a\nname is ending with \u0027.pl\u0027.\nPecan guess it as a perl file and then reject the call (with a 404\nerror).\n\nThe fix is to tell pecan to not guess any content_type from extension by\nsetting guess_content_type_from_ext to False on initialisation.\n\nCloses-Bug: 1987418\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I1bc20b953787cae9b42d12bb5eb3c7a94ed4bbdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ffbb00537104e52541ac7979bbb06c2e65ba682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ffbb00537104e52541ac7979bbb06c2e65ba682"}]},"branch":"refs/heads/master"},"2a312cc81446c9e3ac98e6bc698fb44a82cba2b0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-09 08:41:03.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/40/854240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/854240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/854240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/854240/3"}}},"commit":{"parents":[{"commit":"b350f58f3e7f18cf5c0b54b9c4ec7f9e4daeb4bd","subject":"Fix race condition when comparing dates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b350f58f3e7f18cf5c0b54b9c4ec7f9e4daeb4bd"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-08-23 16:11:39.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-09 08:40:22.000000000","tz":120},"subject":"Do not guess content from extension on API","message":"Do not guess content from extension on API\n\nWhen getting a cron trigger (but also any other object) by name, pecan\ncore will try to guess the content_type from the extension.\nWhile for most name, the guess will fail, it may succeed, e.g. when a\nname is ending with \u0027.pl\u0027.\nPecan guess it as a perl file and then reject the call (with a 404\nerror).\n\nThe fix is to tell pecan to not guess any content_type from extension by\nsetting guess_content_type_from_ext to False on initialisation.\n\nCloses-Bug: 1987418\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I1bc20b953787cae9b42d12bb5eb3c7a94ed4bbdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a312cc81446c9e3ac98e6bc698fb44a82cba2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a312cc81446c9e3ac98e6bc698fb44a82cba2b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
