)]}'
{"id":"openstack%2Fmistral~858172","triplet_id":"openstack%2Fmistral~stable%2Fzed~I88ef08b8cb8ac48f5bcab7873173b3e726cd4f42","project":"openstack/mistral","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"I88ef08b8cb8ac48f5bcab7873173b3e726cd4f42","subject":"Update .gitreview for stable/zed","status":"MERGED","created":"2022-09-16 15:19:54.000000000","updated":"2022-09-18 10:22:41.000000000","submitted":"2022-09-16 19:06:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858172-create-zed","meta_rev_id":"1940a98c008a7e045a3a5ab52151da0d9aa9014a","_number":858172,"virtual_id_number":858172,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 19:06:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-16 18:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-18 10:22:41.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-16 18:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 16:07:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 18:21:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-09-18 10:22:41.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"beb4179d8bd9fe4590ca195b1cf333671829a24e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-16 15:19:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d09da9d55e88d3962721e980339abf24a2a403fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:07:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c87d3828c6dc48e1b9f199a42ebbe3c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3be138c9da4b3a8d30dfba7f70c678 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1de13ecf7ffe4acbaf8bc9968bee0174 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74aa6a09c8b441959961cecde9917860 : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbca2941e7954e1fbd8d558fae97a2a2 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08a3e31d5fe411a9ab6f836662dddd9 : SUCCESS in 10m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e48e0f7f14d46ef8ff7f71416b99b07 : SUCCESS in 10m 43s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1f1f2c308f874fbc99c0fa8bc8ee9be8 : SUCCESS in 39m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/96ad261ed2544a4d97289b25d24d5f62 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9197c55af88b1b49ccf409c7ab2fbbafdbba89f0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 18:21:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b7b524645e213780b0f532f12656557807af93a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 18:21:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9944329e362d1d98af776dcedd27067d4288d4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 19:06:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6a0019dcde04ad392c6e4ac77003781\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30101684f6340f38588dd20ca1dcd33 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94857b608c76469cbdc5d719690a94f6 : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cceb4889a1e43de85dc4af14a45bacf : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1e2f8bc9444ad19b9355d97652b97c : SUCCESS in 6m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2906fa24fa61432289c9801ca2f3bb3b : SUCCESS in 39m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/260066ada322458089b2fb629b704b4c : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f1957243fefe42721e24046fdb3717b88309e8c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 19:06:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"35ab3a50b8ef9036cabc3131670c306abf0e558b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 19:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6db1d38f9ac84d9a8675e7c9727d395f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c36b361280a7429abd328303631304f6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"1940a98c008a7e045a3a5ab52151da0d9aa9014a","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-09-18 10:22:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8b29101b9dbae7de28e02b6067c1d49b7d62ade","revisions":{"c8b29101b9dbae7de28e02b6067c1d49b7d62ade":{"kind":"REWORK","_number":1,"created":"2022-09-16 15:19:54.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/72/858172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/858172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/858172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/858172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/858172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/858172/1"}}},"commit":{"parents":[{"commit":"95e30947d7daf2756efd97cdabacdd6d1508ff6b","subject":"Merge \"Do not guess content from extension on API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95e30947d7daf2756efd97cdabacdd6d1508ff6b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:19:53.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:19:53.000000000","tz":0},"subject":"Update .gitreview for stable/zed","message":"Update .gitreview for stable/zed\n\nChange-Id: I88ef08b8cb8ac48f5bcab7873173b3e726cd4f42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c8b29101b9dbae7de28e02b6067c1d49b7d62ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c8b29101b9dbae7de28e02b6067c1d49b7d62ade"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[]}
