)]}'
{"id":"openstack%2Fmistral~859797","triplet_id":"openstack%2Fmistral~stable%2Fzed~I92f6d73ccfb67a633d1494f9b1c6990439924235","project":"openstack/mistral","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2022-09-29 16:11:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92f6d73ccfb67a633d1494f9b1c6990439924235","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-09-29 13:24:22.000000000","updated":"2022-09-29 16:13:14.000000000","submitted":"2022-09-29 16:11:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859797","cherry_pick_of_change":859812,"cherry_pick_of_patch_set":1,"meta_rev_id":"66199302b882326f9e0a2411b7fbf82aa0e2f446","_number":859797,"virtual_id_number":859797,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 16:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-29 15:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-29 15:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 13:24:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-09-29 13:24:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2022-09-29 13:24:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-09-29 13:24:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-09-29 13:24:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-09-29 15:12:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f405857855c01015547790be59c75bb4b992e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-09-29 13:24:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b306e3350a5644932062af6d6fa2071b89993f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbba994fad8b47cd855156bdf84d635a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127448b8fec94c93bc60e8f3c8bc9bc6 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5149026a1f0d447f9377edb5aa407ca9 : SUCCESS in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e52db0f1ce841e8971700f7b4262e94 : SUCCESS in 11m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4861d0a64c1848f9a2dc5c43096e6206 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399e3d012a944cef8e5583fb29b70cc5 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9f7ea1047e24f25ae0792b122fe6adf : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b30630f1a1774b8c878120cdbbf3dbec : SUCCESS in 11m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fa18c957032d4c1db3d32396579c2c9e : SUCCESS in 42m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f510bc11eba946ea84fc70c203913679 : SUCCESS in 49m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b4d8f0e2f724daea6f8514ee002303c : SUCCESS in 50m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/221be14c103d41e1a102f835168d9ca2 : SUCCESS in 46m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/29d0f21de2c64464a2cdbb69312cd73c : SUCCESS in 51m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5509b5b6ee50439ba3cb6dceca9c1f9f : SUCCESS in 36m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fe501a7d1ed046e3b01f7b26e0d6935f : SUCCESS in 36m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b7acf03f0e4046c4b5011b82110282f6 : FAILURE in 1h 46m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb2ec115490a8cc8615b8350fccb4e1eae9d7b29","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-09-29 15:17:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2004e611a9450bb8136729b53dce050224227ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:17:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ee356e0afe0f5b71f79a9c07f0290c8dea53d71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:11:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6a68f3c15864b93ba1a305a5a5f174e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4bbb144fdd54d06b8775fa3ad893392 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733d970ae59a4b6c869278ee28552476 : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da8b6bf6dde44bb59dad146def2e4e8a : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7e4847bbd54b2fbd00ca0cc1b0de1d : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3e930410c044abb894da0988c778b85 : SUCCESS in 4m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3be9ea8bbd224ebf866dad11e2888426 : SUCCESS in 50m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1ff1c9693b2408b9457887a8db5a317 : SUCCESS in 34m 44s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bd30918e1f040b4a3c08baf4e96abae : SUCCESS in 52m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dba55c78c8b747fd875839f4ebb44912 : SUCCESS in 39m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/76f46c9a27db4baa865bef1a40884c63 : SUCCESS in 48m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/19699531bde94633b74199b94d577903 : SUCCESS in 36m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0183b4a5dfd3422698d309011478b8eb : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6de38a5b6759ccf9511d11e45b907a96184e0038","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:11:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66199302b882326f9e0a2411b7fbf82aa0e2f446","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:13:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb450d5be1694213a31d8fa6ba47f710\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c48fc6c29b4224b8571039894ec083 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b450848a3eb6996a886f31d347f899af16c939a9","revisions":{"b450848a3eb6996a886f31d347f899af16c939a9":{"kind":"REWORK","_number":1,"created":"2022-09-29 13:24:22.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/97/859797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/97/859797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/859797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/859797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/97/859797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/97/859797/1"}}},"commit":{"parents":[{"commit":"041e9a33e7ef05285bcae3395cf8b442bbd394d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/041e9a33e7ef05285bcae3395cf8b442bbd394d3"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-29 08:40:59.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2022-09-29 13:24:22.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I92f6d73ccfb67a633d1494f9b1c6990439924235\n(cherry picked from commit 13c015b3fd86e1373b8bbc9bb469330d67fd292c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b450848a3eb6996a886f31d347f899af16c939a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b450848a3eb6996a886f31d347f899af16c939a9"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
