)]}'
{"id":"openstack%2Fmistral~859812","triplet_id":"openstack%2Fmistral~master~I92f6d73ccfb67a633d1494f9b1c6990439924235","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2022-09-29 12:45:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92f6d73ccfb67a633d1494f9b1c6990439924235","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-09-29 08:42:28.000000000","updated":"2022-09-29 12:47:39.000000000","submitted":"2022-09-29 12:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859812","meta_rev_id":"291526402c20ba765299295a9e5800a7a297840c","_number":859812,"virtual_id_number":859812,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 12:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-29 11:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-29 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 09:22:31.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-09-29 09:22:31.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2022-09-29 09:22:31.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-09-29 09:22:31.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-09-29 09:22:31.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-09-29 10:53:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"650745ae9b54698ca1c4d206705d7848ff7a5f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-09-29 08:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52c0b7cc11a16be3067f59746dcdec8a406afd95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 10:53:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0bab80b9d94866b480f0ddf4695fc5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f629b04b1544acbce1f1fae82f632d : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/255a23669ced40cfb5e1753db5e24407 : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd285feaaf7446c19a49bb15c306b2ee : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc35b922e2cd44a4873567345b082a89 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33ac9a23bca4483a6edee3bfe59e272 : SUCCESS in 8m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2111e947a3c495a88e43838ba368101 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d29c9d09024917ac02456237a5be09 : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a216e65c10e24551b6950d1bbb7f9e75 : SUCCESS in 12m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/31ab8f665d7f4d05a4f5a8fe78376560 : SUCCESS in 48m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11cb420dccaf4ffc85b03725c0974d5d : SUCCESS in 42m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/497dd7d4e3cb48dcbba27fc751bba8f5 : SUCCESS in 40m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ff31ef6b65ea4760bc53fdf323a8dd7f : SUCCESS in 49m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e723b89621474511a8aef82649578041 : SUCCESS in 43m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aaaf76b2af1d4c1096039968026f1255 : SUCCESS in 40m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df0dff53ab8f41d3b6f81ff7af58c2d0 : SUCCESS in 33m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0fa07578dade4157ac90c0c4825e8d33 : TIMED_OUT in 2h 04m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee2e5a7a9aa74c516321c2d2dfcd0e288914d623","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-09-29 11:34:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fde8bd35cf5939605a77c923ae1433eca6502c40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:35:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13912998fa7bc0dd56f9e5d61883cad9b0a53e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 12:45:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f149d415494a10ac510487ac10afe8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcdd4f4e811f4508a2a4ab50381ac68c : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc3ad607abe044bb8fa0031a351b469d : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/774cbe01aa7b44aabef8aab0fcd3d600 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137c4859b9ec49e7b5b7a85e28da1ea4 : SUCCESS in 7m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0ad048e782547e09c84c0611999f0a3 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ed8ec928d224429944c200709f6dccf : SUCCESS in 3m 41s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/40a1aa79b6504963805346e0fc09078e : SUCCESS in 1h 04m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51c0f0ddbfc34d1bbfc1120138b9da0f : SUCCESS in 45m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5d6f5050e244897b8bf46e5008f0c6f : SUCCESS in 53m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a7f9cfb420234556b4512d1009921852 : SUCCESS in 50m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a50391f2bf83413c910a03eed89908b3 : SUCCESS in 50m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/99db05eedca543a78870030a8b23e8fd : SUCCESS in 39m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5d049b5bac424c4288d3889c6b51f769 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"006407997fc84000445671e2f21f4dff9145d27d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 12:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"291526402c20ba765299295a9e5800a7a297840c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 12:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6718fbaada4046e490838052f32f2bd9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f58231c7b4142eabaa5b76f096835c7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13c015b3fd86e1373b8bbc9bb469330d67fd292c","revisions":{"13c015b3fd86e1373b8bbc9bb469330d67fd292c":{"kind":"REWORK","_number":1,"created":"2022-09-29 08:42:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/12/859812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/12/859812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/859812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/859812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/12/859812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/12/859812/1"}}},"commit":{"parents":[{"commit":"5881bcc466a36cbce2ee72170823b6541aef8cfd","subject":"Merge \"Added sorting and marker for pagination of actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5881bcc466a36cbce2ee72170823b6541aef8cfd"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-29 08:40:59.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-29 08:41:56.000000000","tz":120},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I92f6d73ccfb67a633d1494f9b1c6990439924235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/13c015b3fd86e1373b8bbc9bb469330d67fd292c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/13c015b3fd86e1373b8bbc9bb469330d67fd292c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
