)]}'
{"id":"openstack%2Fmistral~860348","triplet_id":"openstack%2Fmistral~stable%2Fxena~I15502818594ce73a265adc841707609c17270fc3","project":"openstack/mistral","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-10-06 08:44:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I15502818594ce73a265adc841707609c17270fc3","subject":"DNM: Pass custom HTTP header by default","status":"ABANDONED","created":"2022-10-05 08:44:36.000000000","updated":"2022-10-06 08:44:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fe5f3e99cb569048b57dd31723f9a70b9f08a2f","_number":860348,"virtual_id_number":860348,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-05 15:00:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 15:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3778f3ef4c8c7c8405db7b0c332d77145fe8ffc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-05 08:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4dfe29c4b6e4093afe92e7632e4780dfd56d525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 15:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fddb4cd888a4358abf70285ea2812b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/970746f315c9468bb4f16214b570b137 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ee6c24df824aa69c11f59cddb0d549 : FAILURE in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/946807789eb3470a85dc14815514b48f : FAILURE in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4475ffb4e0544061b59d25feaae9929a : FAILURE in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c23a93807a4fb0af1273128fc37c46 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df873a46451406181e627fa180536aa : FAILURE in 7m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/68b25f04d11742159d9a38e20ea934dd : SUCCESS in 46m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a58ba16bc814d9b90dc9b0d25b3a80a : SUCCESS in 47m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/424888b38ac34787b4eef00accb0ddb5 : SUCCESS in 42m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b8df86feea814013b67282b4a584d6ff : SUCCESS in 40m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d3da7cb617dc49ba870c01c228a9736f : SUCCESS in 44m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f10bf39118244d6c9922870524fed8c1 : FAILURE in 28m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5d57125f546b4aa9b2131b8f34c310e2 : FAILURE in 25m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/780427b4b31343ce8abeb69329ca16fc : TIMED_OUT in 2h 03m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fe5f3e99cb569048b57dd31723f9a70b9f08a2f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-06 08:44:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ad476f137720c81cc56e5caa2346c4055e30c97","revisions":{"4ad476f137720c81cc56e5caa2346c4055e30c97":{"kind":"REWORK","_number":1,"created":"2022-10-05 08:44:36.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/48/860348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/48/860348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/860348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/860348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/860348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/48/860348/1"}}},"commit":{"parents":[{"commit":"78d225b32d221292c3458a605fb3e4da138e020a","subject":"Update zuul queue configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/78d225b32d221292c3458a605fb3e4da138e020a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-05 08:42:10.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-05 08:44:27.000000000","tz":0},"subject":"DNM: Pass custom HTTP header by default","message":"DNM: Pass custom HTTP header by default\n\nChange-Id: I15502818594ce73a265adc841707609c17270fc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ad476f137720c81cc56e5caa2346c4055e30c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ad476f137720c81cc56e5caa2346c4055e30c97"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
