)]}'
{"id":"openstack%2Fmistral~864080","triplet_id":"openstack%2Fmistral~master~I4c21b6527f37918397f7fb425075eb6c07534237","project":"openstack/mistral","branch":"master","topic":"migrate-to-jammy","attention_set":{},"removed_from_attention_set":{"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-11-16 17:06:47.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-11-16 17:06:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c21b6527f37918397f7fb425075eb6c07534237","subject":"Update package name in devstack for ubuntu/jammy","status":"MERGED","created":"2022-11-09 05:38:34.000000000","updated":"2022-11-16 17:07:54.000000000","submitted":"2022-11-16 17:06:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864080-migrate-to-jammy","meta_rev_id":"4c25bfbaf3ef4fdb72ad9d53b8bd0d6a0d0d0db5","_number":864080,"virtual_id_number":864080,"owner":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 17:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2022-11-16 16:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2022-11-13 04:47:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2022-11-16 16:18:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 07:04:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-13 04:47:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-11-13 04:48:10.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-11-16 16:18:35.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"46c4aec4000d221f6aa29a523349bf497e1c6532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-11-09 05:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"278ac995307a34ff6778465de0b5c4692c6bea0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 07:04:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1356abedc5f54a0f9127f46f4a2c57bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6693dc715595496e99d9dff9258819e4 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/289aa1f593fc4dcba50d1b23b6c0e727 : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39faa916108740678e24373d62304d80 : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0868dee973c44ff1bc529a2d0ce5e3cf : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bda0513cea3465e9c1bdaf128b7c6f3 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07b756706b2244cb94f6b685877c59b9 : SUCCESS in 13m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7cc18d0305c14048b76cfc490ff3e5a6 : SUCCESS in 45m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3f7ffcb4c5b45d486250cbdf514a9c1 : SUCCESS in 44m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bff46a62d331469aad0245c14861a8f7 : SUCCESS in 46m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f5103f91cb65475b9fa7633f3abd4467 : SUCCESS in 45m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2d2e0adb02e9412286b8447b7fddef36 : SUCCESS in 46m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/22de01aaeb91499f81c55889e7b58008 : SUCCESS in 39m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d59a384bc9b541f3a793b4d3a0b6ecad : SUCCESS in 36m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b4dbf9416299479b978e9fa0920c143a : FAILURE in 1h 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed1dc37955106b2df59e5ba6d9fa411aa9fcca29","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-13 04:47:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd96aeb1db7f4804d799a0370d67df50bb34f7c6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-13 04:48:22.000000000","message":"Topic devstack-jammy removed","accounts_in_message":[],"_revision_number":1},{"id":"4cf42270181949bfdd4f3fb4896e7c77924cc742","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-13 04:48:25.000000000","message":"Topic set to migrate-to-jammy","accounts_in_message":[],"_revision_number":1},{"id":"ed3c958fe6c40fcd80efe229785f7800f1c64df7","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-16 16:18:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6287ea25d7ea317b0761836ec5c0032cdaa931e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:19:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"555c69acee74229bfa5c760dc811eed5311f6550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:06:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ea44ddbfb41421b9a3f29bf9d367008\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbec0cd80f9d42619bd4ec9bb21d4fad : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c71c23dfb624ea9a399c280f32f6156 : SUCCESS in 9m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0149519a2b4422893b67e4ca366157 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f1271aaeaa43d9ad2cdfd58117844a : SUCCESS in 7m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/da65e1255b93441291d15ec9cda2dfbd : SUCCESS in 36m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cb11fb5e57447e29cbb697f8c1e20ff : SUCCESS in 31m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/053ab45e79ad4ae28eed156a0a5c82b3 : SUCCESS in 36m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/31e7507e39c34359a93d42fd7d9c36e3 : SUCCESS in 29m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/010b15da562640dcba2a107aa38ca1a6 : SUCCESS in 32m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/19671193da12409a8665b5227b2813d7 : SUCCESS in 36m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d2f0bba535244e7ba1f527710cae856a : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9032faed5b42864b0056b70d016bd7cda8fc15aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:06:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c25bfbaf3ef4fdb72ad9d53b8bd0d6a0d0d0db5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:07:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a297281e1b1480ba8710506f5d3650a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89b1807be7b45529718c19380a5479f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","revisions":{"c820d6f36549f4ece14f0085bc7249ea3bb74a4b":{"kind":"REWORK","_number":1,"created":"2022-11-09 05:38:34.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/80/864080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/864080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/864080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/864080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/864080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/864080/1"}}},"commit":{"parents":[{"commit":"55745a750c1912b9e37f27bfcc72f7bf1ab80159","subject":"Merge \"Remove extra description on cron trigger executions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55745a750c1912b9e37f27bfcc72f7bf1ab80159"}]}],"author":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2022-11-09 05:23:25.000000000","tz":0},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2022-11-09 05:38:31.000000000","tz":0},"subject":"Update package name in devstack for ubuntu/jammy","message":"Update package name in devstack for ubuntu/jammy\n\nFor python3, `python-nose` has been changed to `python3-nose` in Ubuntu\nJammy Jellyfish. This change is to update the package name to avoid an\nerror on exit.\n\nSigned-off-by: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nChange-Id: I4c21b6527f37918397f7fb425075eb6c07534237\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[]}
