)]}'
{"id":"openstack%2Fmistral~865599","triplet_id":"openstack%2Fmistral~master~Ib802a1b54e69c29b4d0361f048c2b9c076a4c176","project":"openstack/mistral","branch":"master","topic":"bp/mistral-task-skipping-feature","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-02-14 11:41:40.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}},"hashtags":[],"change_id":"Ib802a1b54e69c29b4d0361f048c2b9c076a4c176","subject":"Task skipping feature","status":"MERGED","created":"2022-11-24 19:23:24.000000000","updated":"2023-02-14 13:23:18.000000000","submitted":"2023-02-14 13:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865599-bp/mistral-task-skipping-feature","meta_rev_id":"8b544bc07d75a4507872f8435010958b33c831c4","_number":865599,"virtual_id_number":865599,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 13:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"recommended":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"all":[{"value":1,"date":"2023-02-02 11:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},{"value":2,"date":"2023-02-14 11:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},{"value":1,"date":"2023-02-14 11:41:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 21:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 07:38:59.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-12-01 07:38:59.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2022-12-01 07:38:59.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-12-01 07:38:59.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2023-02-02 11:42:20.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"state":"REVIEWER"}],"messages":[{"id":"abb754df17fb8e4810297a69a3de8c218beb148a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-24 19:23:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cb9c931c6741225cb041f93120858fc5ff2b56f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-24 19:35:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5752c36f708b48c0db97599adf9410364af4f4cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-24 19:43:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c2e00bacb61875d1942d9a23ae6fd2db057ce84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-24 19:44:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0acca7ac7bee73b72e46dc83522f259bcd855583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 21:15:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/921a6ed1020b4cf3b8683b8ac507f264\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec443f0d2a04cf088e41f0d5a4b6e58 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3df9a2efe700436d9f50c625031ee506 : FAILURE in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b50588a9f5d40718708129ab2f2a788 : FAILURE in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d3e379670bb491489ae336bb718a67d : FAILURE in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb73065050b74040991b69caf76b444b : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb9a296a4a4d46d6bdc195099d139ee9 : FAILURE in 11m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ed4f70795c904bb084cbaa4ddaf0f83d : SUCCESS in 36m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e1c2d98ddf040e692d5b75bb90f149f : SUCCESS in 35m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddce60830b4c4a8fa2f341edc24465d4 : SUCCESS in 34m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/93fc37c50ae7411d860c6a81ec2ca00c : SUCCESS in 34m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4c3d55ddb76f4cfaa0690f8709c49e7e : SUCCESS in 34m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dfd20c5084c248aba1545ad7d40b3bb2 : FAILURE in 37m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d867bfa6aa464ce2802a13713ae6a77e : FAILURE in 34m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/654de40aa1194eeebf7a2e294dbb3cb4 : FAILURE in 1h 30m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a1a22729cfc12db15d3af9e96c0f5f4660e0d920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-26 14:20:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7f7f0e116b1cb52b1e7772c3cf4e67629c80f800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-26 14:32:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"32e44e6c61ef14cad2b26a9c6f5d41676d91541f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-26 14:56:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3de31ae4a8286728cf1a5a12e001df59762bc593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-26 15:10:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c9823e34de8a293c3502b03907b8ac35cc306f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-26 17:18:24.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0659fc6f89254fa29cda4d7d0a7f2eeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22149419497447b9873f5179cbd66bcd : FAILURE in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b95004645b7b44af9e4545acc81d5213 : FAILURE in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e0a43ed8634467094c2f65e3330e8da : FAILURE in 10m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e044adb14eb7480bbfde2f971e9751b8 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95debe1e076c495cb726b5b1cf684244 : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58a98a61bd3543fc9bbfc6970d62cd05 : FAILURE in 14m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d7b5b9c28d9440a5a3bf408cd28e648a : SUCCESS in 42m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd55d6adc5c140e5996ec7448f5c90c7 : SUCCESS in 40m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b76973db3cb14d069e4229ae72e2eccc : SUCCESS in 47m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c25c1c4497cc469c88e2fd93a820a10d : SUCCESS in 41m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/07254335974348f799148ecf7e891ba6 : SUCCESS in 49m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/482719d2cc7d4ee0aa4cf45880cbef6f : FAILURE in 40m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/87ee6fcfd5d445229e5719856a25fe63 : FAILURE in 39m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/df943e65a8f6458ba0da767e90b83a49 : TIMED_OUT in 2h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"beac8b9da7218125cb01cd8237bb372697e0e425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-26 19:16:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"82f36a7cc8f6ffc1be1e3171bf50fe29b39f8b6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-26 20:58:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c56bfcdd8734f7f8875e2c55f94d44e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5f15e7776444c6aa73ba3e7cd2e1f2 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/466aec15af3c4fc38764ba5e2da80bce : FAILURE in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cf13c1fd85d491d904ca148aef78310 : FAILURE in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98f78cd0c8314ca3b26d469bf391da22 : FAILURE in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440acc4c24e64f14b7148a9a6933f87f : SUCCESS in 7m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9707bf583eaa4fcea5cc243ef2530871 : FAILURE in 11m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8643aa99c0a346f7beb19c6516e43c25 : SUCCESS in 38m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71303cf687f2451aadf25c5bada092f9 : SUCCESS in 32m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7907b477e6cb4ffca04475bf16e86dd4 : SUCCESS in 34m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/142954862cd049dc8506d256e84aab56 : SUCCESS in 32m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/441f4f31f068473686dead19121ff135 : SUCCESS in 35m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c12e58b4c9674dc98946b40600a2bfe2 : FAILURE in 39m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ded950968443423bbae8357102a91481 : FAILURE in 37m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b29db9f8a0d4499387e807f9dd9c6ffb : FAILURE in 1h 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e2a5085d1d2c70f1567bba18651eaecdd15eaab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-26 21:31:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"97987b7acc8ae6e6b93a54c85908801b1a90ea41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-26 23:39:17.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cbd2c4415244217936d348b12245183\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ba065bb6fea4d0f8f23d3aec838976c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a80c44e6486475884c1ffc769fdf25d : FAILURE in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be2d2d9b1c7490da8868eb8385d1bb7 : FAILURE in 10m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a020c6b65114d9fbd09a5c1ade45eb4 : FAILURE in 10m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99909a08b670478fbcbe2838905095a1 : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c504780cf6b4801a15cd76b155b4c2a : FAILURE in 11m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b7e57572ae7474f998371f3dcf4dc35 : SUCCESS in 34m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3fd0036f706441da18078f47d96ade8 : SUCCESS in 33m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cbe90c1707f4ee39560613c5d55e36f : SUCCESS in 33m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/deabe2d763aa4582b8a6a68645b92bcb : SUCCESS in 35m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/23ffb811db494235849daee5ca48082c : SUCCESS in 37m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3792ee28e72c42efbf24cbe58b9004c9 : FAILURE in 37m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b6ac4c932c4a417a9785202d71906499 : FAILURE in 38m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/921ad46265164941886799905075612a : TIMED_OUT in 2h 05m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9d587134733e52d42f003edbad42691623ce8885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-30 16:40:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"20e6813d0dd78f5e25b0c10a15d4b09035a9442a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 18:49:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d75907f7e243cfb81b08748cdaf3ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aef6ee55afd447cb945ad79d3c8b423 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbfb39225d3f4f4b867ac2352236f059 : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4740ed54f14f481a8f70246b285414db : SUCCESS in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f4c3573930c4a25b94a5b10e3785b92 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6e12b3d7fa4b54a637663809d5acac : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf3a7ca65f814d60b58b660fbc13cc79 : SUCCESS in 12m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d264accbfa0344ac957741ab31e69e39 : SUCCESS in 37m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eddd898718a24db8a61d01ae1e712333 : SUCCESS in 36m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3995e35b6f72485ebc01076cf037c30e : SUCCESS in 49m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e96e2bd24d244d27b8394e981d87243d : SUCCESS in 44m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6dfd7ab2e29949ea8ef8c0fbd38cbbfb : SUCCESS in 45m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7c9ba4fcbe9f4a0399d1469f5301ab9a : SUCCESS in 41m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cc6129a8af1e4982be655742210680f3 : SUCCESS in 39m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6c7cc94e803840629ad57b9b65c825b8 : TIMED_OUT in 2h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fc8f0036af99fa50aca9fdf4ff7932f5835b940d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-30 20:35:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3862340c6ef3999087b515bb08431beae135fa79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 22:41:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3bd3fc6692149108058030f772a0fd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463775690a104f129bf59b511f7d8903 : FAILURE in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36d101fea6c043ee952e26a3da78f6e6 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57746982af404669bd30235fd9410535 : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aefb706c34a2460cba0592164b2dc1c8 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba14a077a0244559d574b825b27ec7d : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb2dcbd92a93469bb6570784fb3f9cda : SUCCESS in 11m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3cdd04efca9b4d528c08527f48db9c94 : SUCCESS in 34m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a7362f0e01841238a4e01b5819adb64 : SUCCESS in 38m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3845bdc6a124f0182b5fe8ac5f0e576 : SUCCESS in 33m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a3ee3689958a49838534079d8a8e4a41 : SUCCESS in 38m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e31a15bc82be41db80dd24a0317e40e0 : SUCCESS in 36m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/192222fb392b45de938e15303f348a05 : SUCCESS in 39m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0bd74ed2c24145d48d8db6c10236ff49 : SUCCESS in 37m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/abc44053104947d79fc090c8f851d8bb : TIMED_OUT in 2h 04m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c3e842a65cf23be2b73f868aaec81a798b328898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-30 22:47:42.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7bc24a9df8752a4a9513f89ab83afda0457f5392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 00:28:38.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83f1dd5f98c04aa18c9512282bc6c886\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/362374b36f7642fcacb2917ef46b6170 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19131b170a47426b9cac3806c03a537d : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4b88575d38448d8aa2d685a47e28e9a : SUCCESS in 9m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efe0f9381ed74755b587f72444f4a99d : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e69b0001e404f5b870377f782d8b6e7 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b221fc3e1a49b39c6bcc607824aed3 : SUCCESS in 15m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/57d1df020fed4980b1ec8cd57a2564ad : SUCCESS in 34m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87fbc18966f04616837bb616bd521d7e : SUCCESS in 34m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72b4246990274b209d1860e9a86999bb : SUCCESS in 44m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8d2adb7311f44ff2b78819f7cb2b7172 : SUCCESS in 33m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/47eec5f7f1cc4743a217cc4c2b8879e1 : SUCCESS in 34m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/67dfcf674ec1463bbd101a64bf651b41 : SUCCESS in 35m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/545d8d70c1df4a2385ffde2b214e6549 : SUCCESS in 42m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a4d56f14dbbc41b68b4d2a6bcd51104c : FAILURE in 1h 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ff83701c1b0bcf2026de46b82e3ff2b49bfb7fc6","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-02-02 11:42:20.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"8489e8d658660cc8b69f9db84f3d8d78c07507a7","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-02-14 11:41:40.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"968dc34789ac386536c23e96975fe53f785c542d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 12:19:15.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f176306b008b1870dd27e81925043a1ec95667c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:22:17.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea1e9197fdb4c72b9404a639903ed8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e5b546a02e408f815c0b6ffa344667 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd1f7dfaf6243f7bd123c7cf0045fae : SUCCESS in 12m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8301cad3bf8b492a8a5b9c5e24a5610a : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b191bab8a441e1985315bec8a4c7c6 : SUCCESS in 6m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c7386986531045008b78ceb41af2f786 : SUCCESS in 56m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd2f1dff991a419b8b1443bc7ad07968 : SUCCESS in 56m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b4421cdac542ff91fdff70389be946 : SUCCESS in 44m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9d6ca57c734040a6a824ec386638eb86 : SUCCESS in 34m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e0d31f44ed445a38b2cfb13420a53f1 : SUCCESS in 33m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fbf5332331ae4b9abe766d4f6ae76ed9 : SUCCESS in 39m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0a8131c6f4b5442582a48552a93f8e1e : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":13},{"id":"56c458a80f9164266330fc18c4a7f6e6fd78c4fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:22:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"8b544bc07d75a4507872f8435010958b33c831c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:23:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9db3eb86158412fad3a5ec8773d7e12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514fc9a4443d4185a0e488c07df2abff : SUCCESS in 41s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"e72a4e9a70156d99f5c0272024aceedeccf01768","revisions":{"1f9021ae01ec850ed21ff5b1af69d96523811200":{"kind":"REWORK","_number":1,"created":"2022-11-24 19:23:24.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/1"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f9021ae01ec850ed21ff5b1af69d96523811200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f9021ae01ec850ed21ff5b1af69d96523811200"}]},"branch":"refs/heads/master"},"182a3e8f442b0cd3d9f92839ab0058675be4314a":{"kind":"REWORK","_number":2,"created":"2022-11-24 19:35:18.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/2"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:34:57.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/182a3e8f442b0cd3d9f92839ab0058675be4314a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/182a3e8f442b0cd3d9f92839ab0058675be4314a"}]},"branch":"refs/heads/master"},"28128e0613e9eda060f7fa4ba48e990b9c565cee":{"kind":"REWORK","_number":3,"created":"2022-11-24 19:43:18.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/3"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:43:06.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28128e0613e9eda060f7fa4ba48e990b9c565cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28128e0613e9eda060f7fa4ba48e990b9c565cee"}]},"branch":"refs/heads/master"},"1e746ecdaf19a0b31c48027802b269017c9effcf":{"kind":"REWORK","_number":4,"created":"2022-11-24 19:44:00.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/4"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:43:50.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e746ecdaf19a0b31c48027802b269017c9effcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e746ecdaf19a0b31c48027802b269017c9effcf"}]},"branch":"refs/heads/master"},"e7a2c5950998143bec243a69e7e2919a0332a959":{"kind":"REWORK","_number":5,"created":"2022-11-26 14:20:22.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/5"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-26 14:20:07.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e7a2c5950998143bec243a69e7e2919a0332a959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e7a2c5950998143bec243a69e7e2919a0332a959"}]},"branch":"refs/heads/master"},"f3337005134fc390b5f4772f88811a1cf2518d85":{"kind":"REWORK","_number":6,"created":"2022-11-26 14:32:22.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/6"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-26 14:32:06.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3337005134fc390b5f4772f88811a1cf2518d85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f3337005134fc390b5f4772f88811a1cf2518d85"}]},"branch":"refs/heads/master"},"83c9cb1183e054504d74b3627bce1a28f1135b99":{"kind":"REWORK","_number":7,"created":"2022-11-26 14:56:37.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/7"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-26 14:56:24.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83c9cb1183e054504d74b3627bce1a28f1135b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/83c9cb1183e054504d74b3627bce1a28f1135b99"}]},"branch":"refs/heads/master"},"5fae8ed36f5c22f3362a0fb81ed4899c0d07a7b8":{"kind":"REWORK","_number":8,"created":"2022-11-26 15:10:09.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/8"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-26 15:09:53.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5fae8ed36f5c22f3362a0fb81ed4899c0d07a7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5fae8ed36f5c22f3362a0fb81ed4899c0d07a7b8"}]},"branch":"refs/heads/master"},"fd54aad526ad710dfbe4d31a7666f30a2d2c518d":{"kind":"REWORK","_number":9,"created":"2022-11-26 19:16:36.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/9"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-26 19:16:20.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd54aad526ad710dfbe4d31a7666f30a2d2c518d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd54aad526ad710dfbe4d31a7666f30a2d2c518d"}]},"branch":"refs/heads/master"},"9f2b3bb999b33a79e63b3a201f05652dc7b888fd":{"kind":"REWORK","_number":10,"created":"2022-11-26 21:31:18.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/10"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-26 21:31:08.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f2b3bb999b33a79e63b3a201f05652dc7b888fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f2b3bb999b33a79e63b3a201f05652dc7b888fd"}]},"branch":"refs/heads/master"},"e08fdb960034aa0744f7532cc804b77afff3063f":{"kind":"REWORK","_number":11,"created":"2022-11-30 16:40:52.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/11"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-30 16:40:35.000000000","tz":180},"subject":"WIP: Task skipping feature","message":"WIP: Task skipping feature\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e08fdb960034aa0744f7532cc804b77afff3063f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e08fdb960034aa0744f7532cc804b77afff3063f"}]},"branch":"refs/heads/master"},"449b0a9d464e2ea80f18d2ade83936f60fc570a1":{"kind":"REWORK","_number":12,"created":"2022-11-30 20:35:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/12"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-30 20:25:48.000000000","tz":180},"subject":"Task skipping feature","message":"Task skipping feature\n\nThis patch adds an ability to rerun failed workflow by\nskipping failed tasks. Workflow behavior in skip case could\nbe configured by new fields in task definition:\n* on-skip\n* publish-on-skip\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nImplements: blueprint mistral-task-skipping-feature\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/449b0a9d464e2ea80f18d2ade83936f60fc570a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/449b0a9d464e2ea80f18d2ade83936f60fc570a1"}]},"branch":"refs/heads/master"},"e72a4e9a70156d99f5c0272024aceedeccf01768":{"kind":"REWORK","_number":13,"created":"2022-11-30 22:47:42.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/99/865599/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/99/865599/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/99/865599/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/99/865599/13"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-24 19:22:47.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2022-11-30 22:47:30.000000000","tz":180},"subject":"Task skipping feature","message":"Task skipping feature\n\nThis patch adds an ability to rerun failed workflow by\nskipping failed tasks. Workflow behavior in skip case could\nbe configured by new fields in task definition:\n* on-skip\n* publish-on-skip\n\nChange-Id: Ib802a1b54e69c29b4d0361f048c2b9c076a4c176\nImplements: blueprint mistral-task-skipping-feature\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e72a4e9a70156d99f5c0272024aceedeccf01768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e72a4e9a70156d99f5c0272024aceedeccf01768"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[]}
