)]}'
{"id":"openstack%2Fmistral~865985","triplet_id":"openstack%2Fmistral~master~I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59","project":"openstack/mistral","branch":"master","topic":"bug/1998210","attention_set":{},"removed_from_attention_set":{"35584":{"account":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"last_update":"2023-01-07 17:49:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59","subject":"Update get execution\u0027s tasks functionality","status":"MERGED","created":"2022-11-29 12:48:27.000000000","updated":"2023-01-07 17:50:39.000000000","submitted":"2023-01-07 17:49:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865985-bug/1998210","meta_rev_id":"f2dfc7af61c216393701f475bf53f0153c644ba5","_number":865985,"virtual_id_number":865985,"owner":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-07 17:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2022-12-01 15:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2023-01-07 17:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-29 15:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 15:38:32.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2022-11-30 07:45:18.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2022-11-30 13:57:57.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2022-12-01 15:47:34.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-12-01 15:47:34.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2022-12-01 15:47:34.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-12-01 15:47:34.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"}],"messages":[{"id":"65b8482f53bb4ca879bef1dfc882e13eb981b385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-11-29 12:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4c89599992d31582a955079aa3e377bbe034df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-11-29 13:30:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f9bfcba73c49f3d4679f9bb5d7720dce7f5a24f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 15:13:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59a96deae783409084925ec115f299e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/438288c051a34206b440ccf6998626d4 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f3a599a1b284af788f7d996a17372d7 : SUCCESS in 9m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3be67a118316480f8be15e04d83e218f : SUCCESS in 9m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76419985c1014d53ad5ea1bb77435b59 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ccf5bf2852b4fed80194069b52ae05c : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37b1ea0d72f84c2f926f9ada071e5e80 : SUCCESS in 13m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c84c7dd5305b4c5a8befde7ec605390b : SUCCESS in 49m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/488aa99291c24b478f14367044c9a64b : SUCCESS in 35m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3fd84d65ddf4977882686a1a46f8657 : SUCCESS in 46m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/011a961733f44151a1db7f4c81b72671 : SUCCESS in 41m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7926bf5e3882426e97fb52adc828cee9 : SUCCESS in 45m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1f725537290b49fbbe5e8b70827b5df2 : SUCCESS in 38m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/109200d9b44841c6990fb84c3e99b02f : SUCCESS in 35m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/069a089833a44d1383dc0a334adf3e5c : FAILURE in 1h 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6493a6c138318fe6512536d3c2289fa9126bb44e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-29 15:38:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6b3c11aee0eb7c94fe466c0feeb22ed505c7338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-11-30 06:19:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"942003d11dd224112759d66e7ccb671d92ed32bf","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-11-30 07:45:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"43938c695b0eeae9f4fb81843b901f42aa0dd831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 08:01:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/949ad7f3f5554156aba10eb95eaa63cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89dc312679564774b404b7be3014da37 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dae7dfb9190a42e28571a84ce8328f49 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6bd370e4c47422f8f75dce785703b69 : SUCCESS in 10m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60d76e5f3542459cb09ed0b7cab068f2 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cee2d2c84be4e24a5cfba38717cc972 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23c6b6431a1a46cbb31f6e25eadaafe2 : SUCCESS in 12m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/38108800fc76408f9e79ab34cfe7c766 : SUCCESS in 34m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b60671efc8c648a5a87a8c49993f7333 : SUCCESS in 28m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65092b413ffe4cd0948a9f860e40ee50 : SUCCESS in 28m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1fa3394a988447d2b2c41b8c4402f368 : SUCCESS in 31m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/33d5e57d6fa24eaf8d59d61f3242977a : SUCCESS in 30m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0e0d3e7bbc31413eb0f9a29fb15808d6 : FAILURE in 35m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1cbe508ea78b41c4be0c36b70fc3c7c5 : SUCCESS in 36m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c1fbc81f697f446db816f7430865abaa : FAILURE in 1h 40m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05b13f2af38e084b62ac39a8a19df45b10f5fd46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-11-30 09:34:40.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d4c6bf359937061df3d3ae827b13abcb960e996c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-11-30 10:22:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4d1f78ac9eac4f427d97d8f116f774339fdfef6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 12:33:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/281b8fd88a144b3a8afc5d74f1fc159d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40136d139a71441fb2579164543db00e : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81e96b5d19444c4cb6695ba45f186d54 : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af7723639924b49a2b9e3bfdb0b8c2e : SUCCESS in 9m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dc5e860005a4856b2ce39cda619469e : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3904fdcc3684f15a37ddad45a0b3b33 : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11bda8468ef941b8bfe470fc905f571d : SUCCESS in 13m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5a51f5cc20784cd5846ca0ab81420c3a : SUCCESS in 48m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/681a32b83e9642279873b48d3dd81117 : SUCCESS in 49m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acd73937efd24111ac4ab12d867b8557 : SUCCESS in 51m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9a8b9c3bfbc54038ab7071a0e6071684 : SUCCESS in 41m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/422e9eb96d9f41cda2a4d9509433e0e3 : SUCCESS in 47m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/05023a8d09364dc0b3ae4bc663255285 : SUCCESS in 37m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4819550c7f154fe49efdbce1dbdeeb15 : SUCCESS in 35m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/36c53483316143c6a57408cc87a0ce89 : TIMED_OUT in 2h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ca7191edc67ff48eed49028cf4886b2a82a3f901","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-11-30 13:57:57.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"232c407b831bfc970b160e8fcdac2204e62e5252","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-11-30 15:31:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bc5b2ce6bba105a4545c9e1a2e75bcca41433059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-12-01 07:39:48.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b60ef22949583f097a350867512723a755f1d1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 09:48:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/559716fad1074d70bb240757325f2014\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b223cb0f6b486a9544ae5c89685c84 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a769256f9c47a1bb177922a1f8afab : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07f8fc5e0e2a4a928f05142ee9187ccf : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02309eabede44a71a85dfa66e60a521a : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c767e7deef846678cddf442e65b3940 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbb9bbc1092a4d86baa204fae557566a : SUCCESS in 13m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/471e28c31b714785bfc7556729f7ab60 : SUCCESS in 31m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/990c41b101c24bfba6b67d323f4bb55b : SUCCESS in 30m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1c6ba8c4c9e4a58bc3b49802f82db22 : SUCCESS in 47m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ba5d36fc672c42fbb338411f47c25799 : SUCCESS in 26m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9af0c47a12174afb9cd80f19db2a005f : SUCCESS in 26m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1924b91b46574f1c99b142653589b5af : SUCCESS in 39m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/305e029512bf4055b7f3c3b4ce12f63d : FAILURE in 32m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f29cc1a906ad40b8be4571e9a64acf0f : TIMED_OUT in 2h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dce2c25a712dd2feef6a638179d0e00c9872b93e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-12-01 09:53:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5ea61dbbcdcdab5e20834643229cae18132143a2","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-12-01 09:55:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b1bf25f5ec4bf17c6cd3c7428c58490e090836ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 12:05:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1f48a891414180a489a86d9236b528\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd34ab4018b47f281bf0f374dd77f90 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4e966ccc55c44beaddaf7e751317490 : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc7ca96138904f31b1562423ae4089c4 : SUCCESS in 9m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b7efe9f119648e8b310864910ed3bc4 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89910797bcf8413b9a2322c33591a97a : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92e7545d464c435ca8f53dbff7cbd29e : SUCCESS in 13m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8fe4894e4ef0430b8b66c20d5520dde7 : SUCCESS in 42m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe5af9cf04a54ecdaac1322514b16c36 : SUCCESS in 47m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ee606b712734ba49506b9dc176b0606 : SUCCESS in 27m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c95bb8190449499d99b094411ebb7b71 : SUCCESS in 49m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d3066f23d4674c43915e23e15fad6afa : SUCCESS in 48m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d0a2b688a9794284bb7fa090a4443943 : SUCCESS in 32m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/98637c396f554bfa85af6313ccc1071e : SUCCESS in 30m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/59c3cc11a54140ccbefa2dd320eac6c4 : TIMED_OUT in 2h 04m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"42bf08990f8f2fc0af85844d7b6d2eec40ef744a","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2022-12-01 15:46:20.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fb4610c56c976d131b41a1c8e984f5fece8feafd","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-12-01 15:46:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c58e4203b6e79d270663cbbdf66e435da40d166f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-01-07 17:16:15.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"751dc57b493bcbdf4b192d2211d4f0173eefbd45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 17:16:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"abaf206da16f50629a1755a761619316454b693b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 17:49:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a2d704010b24597b835687c10fa3d11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b2ac76385a40668552edf5d5ca40ea : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3743d2b65614a50a8bf6a0969384db6 : SUCCESS in 9m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29c134495bf64375b3ce3c9eb4a087fb : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4f490902344feb9fe27234c59e4cd4 : SUCCESS in 6m 42s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3040be08592642eb898052b22b4ed145 : SUCCESS in 27m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b517024f60384d0fadf0c0cdcc855764 : SUCCESS in 25m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00f5a83a545f4d63a1fb466ad4d1987f : SUCCESS in 26m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5506b46a31be4468a541a4733071f1f5 : SUCCESS in 25m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ad6dff93cd7d4c0d98ea1379d1e99b03 : SUCCESS in 25m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4c1a63ebefc6479e9d1249e568513a3d : SUCCESS in 31m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/463fe59559dc468fa9bf6b95396e33c5 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":6},{"id":"d4f00d0bfe6a409976532c9f402ab7914eb6a0c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 17:49:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f2dfc7af61c216393701f475bf53f0153c644ba5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 17:50:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/059f5cdd9d644ae38c321de6bda02882\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d8e99171e549aca3211e87a16e7165 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"87c08ff42172c9fe7735777c98d07ec32c4cd179","revisions":{"2ae3ce8efc542119e9b47b787e9c9e8f74fba3b0":{"kind":"REWORK","_number":1,"created":"2022-11-29 12:48:27.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/85/865985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/865985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/865985/1"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"subject":"Added validation to check if execution-id exists while fetching tasks for execution.","message":"Added validation to check if execution-id exists while fetching tasks for execution.\n\nChange-Id: I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ae3ce8efc542119e9b47b787e9c9e8f74fba3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ae3ce8efc542119e9b47b787e9c9e8f74fba3b0"}]},"branch":"refs/heads/master"},"93a5abca459390587f3614e19687392792a17f14":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-29 13:30:41.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/85/865985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/865985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/865985/2"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 13:30:41.000000000","tz":0},"subject":"Update get execution\u0027s tasks API","message":"Update get execution\u0027s tasks API\n\nthis patch will validate execution-id exists in the DB while fetching tasks for the execution-id.\n\nChange-Id: I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/93a5abca459390587f3614e19687392792a17f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/93a5abca459390587f3614e19687392792a17f14"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f509ed6115c4ecbf56a807a652072949807c9b48":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-30 06:19:00.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/85/865985/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/865985/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/865985/3"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-30 06:19:00.000000000","tz":0},"subject":"Update get execution\u0027s tasks functionality","message":"Update get execution\u0027s tasks functionality\n\nthis patch will validate execution-id exists in the DB while fetching tasks for the execution-id.\n\nChange-Id: I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f509ed6115c4ecbf56a807a652072949807c9b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f509ed6115c4ecbf56a807a652072949807c9b48"}]},"branch":"refs/heads/master","description":"Edit commit message"},"33bc3beeb3de7c4dfbe40a513ba098b0103a8a35":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-11-30 09:34:40.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/85/865985/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/865985/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/865985/4"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-30 09:34:40.000000000","tz":0},"subject":"Update get execution\u0027s tasks functionality","message":"Update get execution\u0027s tasks functionality\n\nAfter this patch, it will validate execution-id exists in the DB while fetching tasks for the execution-id So, GET /v2/executions/execution_id/tasks API will return 404 for absent execution id.\n\nCloses-Bug: #1998210\nChange-Id: I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33bc3beeb3de7c4dfbe40a513ba098b0103a8a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33bc3beeb3de7c4dfbe40a513ba098b0103a8a35"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1f2e0690326eecf2d336083eb5f0c1d3f46fef7a":{"kind":"REWORK","_number":5,"created":"2022-11-30 10:22:29.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/85/865985/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/865985/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/865985/5"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-30 10:21:48.000000000","tz":330},"subject":"Added validation to check if execution-id exists while fetching tasks for execution.","message":"Added validation to check if execution-id exists while fetching tasks for execution.\n\nChange-Id: I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f2e0690326eecf2d336083eb5f0c1d3f46fef7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f2e0690326eecf2d336083eb5f0c1d3f46fef7a"}]},"branch":"refs/heads/master"},"87c08ff42172c9fe7735777c98d07ec32c4cd179":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-12-01 07:39:48.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/85/865985/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/865985/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/865985/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/865985/6"}}},"commit":{"parents":[{"commit":"c820d6f36549f4ece14f0085bc7249ea3bb74a4b","subject":"Update package name in devstack for ubuntu/jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c820d6f36549f4ece14f0085bc7249ea3bb74a4b"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-11-29 12:46:56.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-01 07:39:48.000000000","tz":0},"subject":"Update get execution\u0027s tasks functionality","message":"Update get execution\u0027s tasks functionality\n\nAfter this patch, it will validate execution-id exists in the DB while fetching tasks for the execution-id So, GET /v2/executions/execution_id/tasks API will return 404 for absent execution id.\n\nCloses-Bug: #1998210\nChange-Id: I6ac05f4b87c470a68cd67f19d1bd95a4d6cebe59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87c08ff42172c9fe7735777c98d07ec32c4cd179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87c08ff42172c9fe7735777c98d07ec32c4cd179"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[]}
