)]}'
{"id":"openstack%2Fmistral~868869","triplet_id":"openstack%2Fmistral~master~I48859d5a1c8242a6a7d049c96f8dcabbf4c7c044","project":"openstack/mistral","branch":"master","topic":"bug/2000727","attention_set":{},"removed_from_attention_set":{"35584":{"account":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"last_update":"2023-01-03 12:40:26.000000000","reason":"Change was submitted"},"21970":{"account":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"last_update":"2023-01-03 12:40:26.000000000","reason":"Change was submitted"},"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-01-03 12:40:26.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2023-01-03 12:40:26.000000000","reason":"Change was submitted"},"8731":{"account":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"last_update":"2023-01-03 12:40:26.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2023-01-03 12:40:26.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-01-03 11:58:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I48859d5a1c8242a6a7d049c96f8dcabbf4c7c044","subject":"Update tox.ini to work with tox 4","status":"MERGED","created":"2022-12-29 12:59:44.000000000","updated":"2023-01-03 12:41:29.000000000","submitted":"2023-01-03 12:40:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868869-bug/2000727","meta_rev_id":"9fe06c45d40c3d8144e16de8e90bea8f7965df1d","_number":868869,"virtual_id_number":868869,"owner":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-03 12:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-03 11:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-03 11:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-29 14:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 10:53:39.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2023-01-02 10:53:39.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2023-01-02 10:53:39.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2023-01-02 10:53:39.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2023-01-02 10:53:39.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2023-01-02 10:53:39.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"eaa1673ece68c9d9d04742976c78c6f6ed00fcb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-12-29 12:59:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c25d81258c8054783ad0d97aa5b0d1659e7d646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-12-29 13:04:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"417769e7582c1ba441f07ae27e9b7c5dd98a12d0","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2022-12-29 13:08:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8c4101a5a2fb07efebe170c6791c9512d0862d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 14:59:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d0878452e7545c2ad84abe2cc3fc447\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c2e121fed1478dbd10d86565d0e113 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cbed89267224f328f45d66c82561345 : FAILURE in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bf3be8add0a44bfbfdaf1dabcdf9e92 : FAILURE in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef616b9a55e04246af4a154d66ba4532 : FAILURE in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03cef8d2703a41da9a923386d6a7408c : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fd37cb4668e4f86a85d6bac22059ae4 : FAILURE in 4m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f623e03a7a0d4490b599e5d42c400bea : SUCCESS in 33m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddd5942c82fe471c94f034b934978660 : SUCCESS in 24m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f21f2ce31d3244fd971fee5823a46f77 : SUCCESS in 43m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/def4584175bb4fa29c2dc0395a80e643 : SUCCESS in 32m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c0af68e265784e488cd8c89e7ef65bb3 : SUCCESS in 25m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/81d0f3122f8b4afd98d19afea649c4c9 : FAILURE in 3m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/377dec1de8204ae3bdcc7af0fb580a99 : FAILURE in 3m 40s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/83409131adf142498c15a5893b18bdc7 : SUCCESS in 1h 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e92229b1bacb2be94727d6c19dcee5f1bf10d09d","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-02 07:47:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fe96df702e62ab6b9c6de5eb97bfa27dafb3870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-02 08:07:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c245de7ed04fd201efee98dcfe8b7a33589667b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-02 08:29:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e6253c518b00eb5ba77efa0aac5190c72007170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:39:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de871c63b182415fb749618f66616411\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374cdc16623e40de8638d53381efb7ce : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d8dea5bd1fc4852bd542e231c283432 : SUCCESS in 9m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7815220d6d21434bac525b551af40663 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5783f0a876ae4587bdfc383ed9dbb622 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/837bd5f2537f4ddf9156d287553a6f6e : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab5d949761464b0e9bb2ccb86affa777 : SUCCESS in 13m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f032dc361a34e7ba1136846d0ad262d : SUCCESS in 45m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/884675b8463741a585b59594f8300400 : SUCCESS in 48m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52a1ff10ed7b4ee292d9ca38a1445fb7 : SUCCESS in 46m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/04dcc2c6946b4ddd9055e96d75648a5b : SUCCESS in 45m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e40afc3163464a2cbead5384b9c02f0c : SUCCESS in 47m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/618b5f1e44c445db8bb36dbd2aed00b5 : SUCCESS in 39m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dbfd795fafa24597aaf558889e243251 : SUCCESS in 37m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/729cd28a28f54133ad8cfc348714c166 : TIMED_OUT in 2h 02m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5161b1b7b0e56d669f4ccd022a66d8dd66d6dde3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-03 11:58:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"922cc7bf1221a8bae768b39d36ee1b084da85ab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 11:59:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e5a2e390218dfb158f5ebab3695f190b2bb808fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 12:40:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5f5b86a6ae4ba2a2324a98840b1904\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8283f2fd2fe41089bf450eceec41305 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8860a0792ce3473ab2e3e7171d440ca4 : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e81538a112d4a77bdd4de20eff49c55 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1399f4e33b40c2984c3c70cbe99ac7 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28da9cc0421c4103b6db7a607c63e3ce : SUCCESS in 3m 22s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6a02e667be2e40d682b19a59f314d637 : SUCCESS in 31m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caa209782fd347b3ab16b2fc0dfdb091 : SUCCESS in 33m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5149780937914cb998b374c272d6676f : SUCCESS in 31m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ca98374ea2824730a8a0e6b00de608c5 : SUCCESS in 25m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b3184f93d0a2455ab8b7c98349415f55 : SUCCESS in 28m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/109ace79568e4726b4f691ef0e6bd48c : SUCCESS in 36m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c098c4d7fd6146db8725d4504ac48c77 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":4},{"id":"691aacf18fd64fc211e2d2be48d65efdeb7638c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 12:40:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9fe06c45d40c3d8144e16de8e90bea8f7965df1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 12:41:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8fd6d4f5ca04e5586eac8b7f6ee6d42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7d223db7114e24a07ccc2286369a53 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3374732e9d9147a896681aeb956867d0 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b30ff6f95525dc1dbdba1642ced07e008f58df14","revisions":{"0de9f0c8e342e2aada4657f3ce5768e607060184":{"kind":"REWORK","_number":1,"created":"2022-12-29 12:59:44.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/69/868869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/868869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/868869/1"}}},"commit":{"parents":[{"commit":"d035624153011673e8d542a7c88a8f717f46475a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d035624153011673e8d542a7c88a8f717f46475a"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-29 12:13:45.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-29 12:59:00.000000000","tz":330},"subject":"Update tox.ini to work with tox 4","message":"Update tox.ini to work with tox 4\n\nThis change will allow this repo to pass zuul now\nthat this has merged:\nhttps://review.opendev.org/c/zuul/zuul-jobs/+/866943\n\nTox 4 deprecated whitelist_externals.\nReplace whitelist_externals with allowlist_externals\nTox 4 Fail on pass_env/passenv entries containing whitespace.\nUpdated passenv with comma separated values\nCloses-Bug: #2000727\nChange-Id: I1750d7286c302436dabb803019bd98f8d60f289f\n\nChange-Id: I48859d5a1c8242a6a7d049c96f8dcabbf4c7c044\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de9f0c8e342e2aada4657f3ce5768e607060184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0de9f0c8e342e2aada4657f3ce5768e607060184"}]},"branch":"refs/heads/master"},"d7045dc90596d9f15e9bebce6be08dd13fdd4116":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-29 13:04:26.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/69/868869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/868869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/868869/2"}}},"commit":{"parents":[{"commit":"d035624153011673e8d542a7c88a8f717f46475a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d035624153011673e8d542a7c88a8f717f46475a"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-29 12:13:45.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-29 13:04:26.000000000","tz":0},"subject":"Update tox.ini to work with tox 4","message":"Update tox.ini to work with tox 4\n\nThis change will allow this repo to pass zuul now\nthat this has merged:\nhttps://review.opendev.org/c/zuul/zuul-jobs/+/866943\n\nTox 4 deprecated whitelist_externals.\nReplace whitelist_externals with allowlist_externals\nTox 4 Fail on pass_env/passenv entries containing whitespace.\nUpdated passenv with comma separated values\nCloses-Bug: #2000727\n\n\nChange-Id: I48859d5a1c8242a6a7d049c96f8dcabbf4c7c044\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d7045dc90596d9f15e9bebce6be08dd13fdd4116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d7045dc90596d9f15e9bebce6be08dd13fdd4116"}]},"branch":"refs/heads/master","description":"Edit commit message"},"db95caf5f5fd08a87f7eb9a12f3c663114da1bf6":{"kind":"REWORK","_number":3,"created":"2023-01-02 08:07:54.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/69/868869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/868869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/868869/3"}}},"commit":{"parents":[{"commit":"d035624153011673e8d542a7c88a8f717f46475a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d035624153011673e8d542a7c88a8f717f46475a"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-29 12:13:45.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-02 08:07:03.000000000","tz":330},"subject":"Update tox.ini to work with tox 4","message":"Update tox.ini to work with tox 4\n\nThis change will allow this repo to pass zuul now\nthat this has merged:\nhttps://review.opendev.org/c/zuul/zuul-jobs/+/866943\n\nTox 4 deprecated whitelist_externals.\nReplace whitelist_externals with allowlist_externals\nTox 4 Fail on pass_env/passenv entries containing whitespace.\nUpdated passenv with comma separated values\nCloses-Bug: #2000727\n\nChange-Id: I48859d5a1c8242a6a7d049c96f8dcabbf4c7c044\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/db95caf5f5fd08a87f7eb9a12f3c663114da1bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/db95caf5f5fd08a87f7eb9a12f3c663114da1bf6"}]},"branch":"refs/heads/master"},"b30ff6f95525dc1dbdba1642ced07e008f58df14":{"kind":"REWORK","_number":4,"created":"2023-01-02 08:29:08.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/69/868869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/868869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/868869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/868869/4"}}},"commit":{"parents":[{"commit":"d035624153011673e8d542a7c88a8f717f46475a","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d035624153011673e8d542a7c88a8f717f46475a"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2022-12-29 12:13:45.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-02 08:28:55.000000000","tz":330},"subject":"Update tox.ini to work with tox 4","message":"Update tox.ini to work with tox 4\n\nThis change will allow this repo to pass zuul now\nthat this has merged:\nhttps://review.opendev.org/c/zuul/zuul-jobs/+/866943\n\nTox 4 deprecated whitelist_externals.\nReplace whitelist_externals with allowlist_externals\nTox 4 Fail on pass_env/passenv entries containing whitespace.\nUpdated passenv with comma separated values\nCloses-Bug: #2000727\n\nChange-Id: I48859d5a1c8242a6a7d049c96f8dcabbf4c7c044\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[]}
