)]}'
{"id":"openstack%2Fmistral~869193","triplet_id":"openstack%2Fmistral~master~I0b7202589eee68ba5560bf2aa60fbbd6118f3719","project":"openstack/mistral","branch":"master","topic":"bp/distribute-mistral-operations","attention_set":{},"removed_from_attention_set":{"35584":{"account":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"last_update":"2023-05-17 09:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b7202589eee68ba5560bf2aa60fbbd6118f3719","subject":"Starting tasks via RPC","status":"MERGED","created":"2023-01-04 11:14:39.000000000","updated":"2023-05-17 09:21:39.000000000","submitted":"2023-05-17 09:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869193-bp/distribute-mistral-operations","meta_rev_id":"bb41116f57aebc0d15c28a0e70e873c7d8c6224c","_number":869193,"virtual_id_number":869193,"owner":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 09:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2023-05-17 08:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2023-05-17 08:28:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 12:32:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 10:20:35.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2023-01-13 11:26:50.000000000","updated_by":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"}],"messages":[{"id":"f6eb2ab5020cc20ff7c5af0547a927a0c4bf8289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-04 11:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"808acae77394369106d8132fb6b48d6403db5296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-04 11:49:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ebb5cb1eb18fb63878b772acaefdcb6b4e589fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:32:09.000000000","message":"Patch Set 2: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a14706db3a9044b6b7afaaecd966d09a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2517e7ccac43a9beac94014f9bd911 : FAILURE in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/787325de296d441fbe7d8b5d989f08ed : FAILURE in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d030f13511ad498fb294e12aed0662ad : FAILURE in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f1b73077fd48dfbf2edb8baa8591ca : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cd8374e534047f0aeaa13dbd3e63d7b : FAILURE in 8m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5c4b9d34fb044f108810707394354710 : FAILURE in 23m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13abe412b55e475da4ec363c49011938 : FAILURE in 26m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5da36eb802744ebc980ab08f68400c56 : FAILURE in 35m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/131668657a1047e6b0e04ac73d4bea4e : FAILURE in 36m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/932a0093373442f5b0e6ab7d6cae34ca : FAILURE in 25m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3069f6f36a87431d972331d2832f671d : FAILURE in 12m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d5fa7abc48d14e46bbe5acc47daefd9b : FAILURE in 11m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6b9106ee962b4cf7968d167b956c84c1 : FAILURE in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c485e382c6d6cd6613dd5d96c77fdcbbf0660668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-04 13:01:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9cac8e17491bed2fa11d2a9fe2855875e1de46a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-04 13:11:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8d9944596c9588256ea3e0876ae5bb1def8ebd6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 15:09:38.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bca20dfdf9549e694b78afa36b14527\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca774c63cf8649ebbad090002232ee15 : FAILURE in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcfb2dc56496486fb585304e6808f583 : FAILURE in 9m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0441f8593dda4087935a57452ae34601 : FAILURE in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7bb6d8e491e461ca314cc1ec8cd73a9 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c263fc9f832c4394a7a96c0c4b747656 : FAILURE in 11m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b573559e6b644199de8d40895e03ec0 : FAILURE in 39m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/643a8c162f0045eb8b3713c109f3f45c : FAILURE in 29m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18d434c87a084760b98edfc647d0e374 : FAILURE in 40m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2e04e855622844eaa22fbf7233c0215a : FAILURE in 30m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c531c341e06b475e886d9bbf5327ac20 : FAILURE in 32m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/119548f3d94f44df854e53d020bd8ade : FAILURE in 36m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7de91c170dbe4ef9ad451fd9e7ac8261 : FAILURE in 33m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5584ba635016468cb1b59880a8bcc197 : FAILURE in 1h 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ead0bbc80ced3b8fdea3f6308adf660f29dcf9d1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-05 10:20:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b08ba724cb35477b76f43ee96311e3280da44bee","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-05 10:21:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ba05c167a555840e81ab536d4b5b5db21d9160e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-05 10:34:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1e075a4c627e727ca073169c67540288c71bd8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:45:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e64e91c368647b0acd36368f96f5600\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea035b674034b058bb4d00513922d74 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c04733fa89d74c8d935817d38c467b83 : FAILURE in 8m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a9984b4cb96429fa16bceee3a0b8e61 : FAILURE in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a223a6bb172469fa76fa4ab75d6f487 : SUCCESS in 9m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/030e6d2f24cc4125b091438da638e8b7 : FAILURE in 13m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/42195a4d61a2470c8510939f156dcb9e : FAILURE in 51m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50c767d62c224dc387d233aa9928a77a : FAILURE in 53m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b80d9b0bed7477697ffcc0e07b2bd7b : FAILURE in 46m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ee2dc94087d347e28401b0407b8b1ab3 : FAILURE in 46m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/749f10c3f1c3423a92bb61af0d2ed745 : FAILURE in 49m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/31ed896048d548629beeb13b0f80041a : FAILURE in 40m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/66e9eb6383e34f088eaed51d1d3f9794 : FAILURE in 37m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1d8f06cc6ccd40268472e68101914246 : TIMED_OUT in 2h 03m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad129eeef88ca4aaac17e5fcfce25991322e1812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-06 11:25:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"860ec3098ea272545dc622d3cb2ed4a0cc50c909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-06 11:27:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"74fc2ad53a21df5a12c18266fcdc0a000198687a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-06 12:08:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"240f5164b9f9304091779b1b44a21f6f78cd7289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-06 13:04:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0b02fb961d970be2a6f2344bd4a42b68ff611d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 15:02:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1005d89ff80b4fedb2ff16b0931f8bc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d734eca4243c4e7b806f8406cd925723 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fa4e7fb71ba472294f47a56dced5941 : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4a655f94ef740ddabc54182a45d2a72 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b76e0a24cf4541bc19502707b50877 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7ecaa5308be4a61b556ded7d5849646 : SUCCESS in 13m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b55d6bb3157048a496c473376ddd1643 : FAILURE in 44m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4e03f7f695745a5bb9a4e34fc61fafb : FAILURE in 35m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7df39aeb16a441d9e3832683eb78ba7 : FAILURE in 28m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f30e2ce2cc614cc28f02f2eb3408b009 : FAILURE in 43m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8a8b429d70d4488dbbee93c80193ee10 : FAILURE in 37m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/25d67b5e591a4d579e3bdc5d236fa926 : SUCCESS in 39m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d97cfff6035e4127976000d449bcce5e : SUCCESS in 36m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5c8f0e8ac15b4946868af22d6483b4cd : FAILURE in 1h 55m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3d6f102ff66ea2bae459de80aebc5338eaef4a7d","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-09 06:41:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"241ed49658f7f94392ad3f1cc28372d171c7061c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 08:49:47.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31ac4a6afc46453e8654624e7378decd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538e9f3351214d37894bae729af7f9d9 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99dfa63b8674427db49fc3c241de4e9f : SUCCESS in 9m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6f8a2ac72a64449bc076f1de6f57590 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496e4288f11b45189dac8ed35b307ee0 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e45872e0dbf24cccbca8e9e2df1313d8 : SUCCESS in 10m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/17dcbe90b8c14b1fb27c07323a3348a1 : FAILURE in 34m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0451740152ec417c88734213f440ec1c : FAILURE in 26m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/177df38d525843aea6804929aa24249b : FAILURE in 33m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/693430259bcc47febc776de5ed54c074 : FAILURE in 33m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b21f94a62c414c7589038ce772bb3b72 : FAILURE in 45m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c3329ba3e35942b7a082da3c505ced0a : SUCCESS in 39m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/337539fa6e7b460b8306a0c38d9365a2 : SUCCESS in 30m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4fe4ee896059494a9b8b6fde2b8c0d9b : TIMED_OUT in 2h 02m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e6717e429d83b60d57ccfff32b91325806c631de","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-10 03:32:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"542a130d3a6f15846c96c7395e3b77871ab22db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 05:29:45.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4a8359be2e94523ab249eb7c3f360ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd6b7426cde4a64a7c8caad4f20f786 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4a6653ac03342ecb45eed70a27211df : SUCCESS in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed5b65f2f030488b83972626f9695d27 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d245cb84ef7f470ead3c06d0d1c91932 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c32b145a0b4448d96945a244220807d : SUCCESS in 13m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/838c88ec60c943c5aa8282e2856e3953 : SUCCESS in 30m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a42770e4cff4ba0b695d6c7d85dbde6 : FAILURE in 32m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/455f2b70803a47a68c9756defc8348fb : FAILURE in 29m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c44f62f671f343009651f39f11e50aa3 : FAILURE in 33m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/126592740a60445889f00f44a215300e : FAILURE in 33m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a35f68b17c164a27a72ea25747a985c4 : SUCCESS in 35m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/25296fc6ee3d45f296dae096ec105731 : SUCCESS in 33m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c714844e9f1e4de59aa4dc522e355804 : FAILURE in 1h 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"108e94dfa1948ddf784b86d810b808310967d254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-10 06:54:48.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"6a41d4337f48ac75aec2abb0f76f9f9dd07a9ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 09:01:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/320614650f6447d1b393145f45ce0abf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938387364a48403996f5e3aa3ca26c77 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00906ecab3f442b696f50d272164ca3b : SUCCESS in 9m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/240479210c464171969144a2d9a4677a : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b196af105444c73b0e84b09f8878053 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34d683926d8c4363b95e396cbec12022 : SUCCESS in 14m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/acc5d9292f82487a903f64a1fa9d42b6 : FAILURE in 40m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8ae8e6a77874652a1dc324eb6812149 : FAILURE in 41m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1acd7525a00a44f7b3aba61587357905 : FAILURE in 33m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b0d623b570544659980376974e3182de : FAILURE in 34m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bf34784e90b741f1b0e8ac28aff08783 : FAILURE in 40m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/531508769a1d485d86454d62e198c8e9 : SUCCESS in 36m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/66fd5e22c3ee4693bb15c693f72701ea : SUCCESS in 36m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e0ab21f0323f4c37871a7ff98e8acbcc : TIMED_OUT in 2h 04m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7d753fec42609f0d9bc933ba175c28faf812f175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-11 07:39:28.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"3abd6e390e3fff0c0862ec767ddc1adbc8fcd7a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 09:49:23.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/854c2f8aa8d947f1b4d85e22376021a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35392101e39549ec9463b0035c371e47 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f323c7d0e554945bd980edc0a6f7ee1 : SUCCESS in 10m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce101fd80e044a4b8786b68f7b3f5423 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df76ebc411e44e55bd5bac62f8c94b1b : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed8ce86a31154575abd584c6a652944d : SUCCESS in 12m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/444fbbfcab674cdbb9038ebd9e1664a5 : FAILURE in 32m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adb19fec95304053ba051c1346b30c04 : FAILURE in 42m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764b908293a44590a3f01a0e341de5db : FAILURE in 33m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/529d190acba54e63b4536f380d3c9444 : FAILURE in 41m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1c9ce93a6cbc475287f4404ee0ebc6d6 : FAILURE in 36m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/40ef8232bfdf405797813c6c6d7ec0ab : SUCCESS in 37m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/69d05a578f764b119da5dc5781535384 : SUCCESS in 38m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0a881d61404549ef8c19b000e0cbf69f : TIMED_OUT in 2h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d2350659361fc6419cbecc0bb7fcc7128e21f07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-13 11:24:43.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"4fa349f1491544eedcef63e8b981831a37603c01","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-13 11:25:57.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3e2fd4faf8b0929c13908a7f72d6ec08503c2e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:29:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a781637c6bdb4f4faebb5232f72b78bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5532805e8d42679382cdbeb95ce76b : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56eb58a51e234b55b219b2fd332b5044 : SUCCESS in 10m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7aecbae9df04a95b087194a9dcb820d : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33872dae57b4dc599d01c338404e1bf : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff40479d802d4377833faf8a4e193aa0 : SUCCESS in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/caec6fcccc954905a09bf432c79a30b4 : FAILURE in 37m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2750083663f44baab2689a531be5214 : FAILURE in 32m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a3e88b427d44d19aec7f148806437f6 : FAILURE in 38m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fbc2ec3140a74d22b874150bb240f6f3 : FAILURE in 34m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6adf66b7f7834673a699f0eae9b1dd7c : FAILURE in 45m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/52b47325ef88481c81f34be0f3ae210d : FAILURE in 38m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1eba54acd5744b109295e391650e1469 : SUCCESS in 30m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f38fce987a934e4ba62a9a519a405f0f : TIMED_OUT in 2h 02m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"40ade7bde1a6bb48d7cdf401ff0cfd4ec9ee0961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-01-19 05:52:44.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"5129074922f9477955ec1316f473fd8a1abea20a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 07:52:59.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97d40a4e67d4d58a30799e1ed4671bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e78f0c6557492ca6810b1d569cc139 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a501092b1a240b4939f53043c52a92e : SUCCESS in 8m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6649bab57904a93a6fdaee3129ad543 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30259716d1442cca02c219571c24cd9 : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab05f00a2f19414d9289517aa2de3efc : SUCCESS in 10m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/215992a9f2674f3cbb45a87ab7110d88 : SUCCESS in 31m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2c23eb73e724864aeccb983a591fdb0 : SUCCESS in 31m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df8edc70d14b4869aec0c8c56032ad75 : SUCCESS in 31m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b7fccb94494143a0859be35acc354eba : SUCCESS in 31m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/41af0a699681411d9d575fb94b5c268c : SUCCESS in 34m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/369c78e9ac464a0aa8a4aa5ccd69c38b : SUCCESS in 36m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0c73c6145dd74442902b4b09d12f8a60 : SUCCESS in 31m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dd827d56551445b4baaa0b1ed476061e : FAILURE in 1h 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"659070c6f16087ed6964e66a9f64c84bdb587f66","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-01-25 12:53:54.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"3200beb8ab2d18a62606bb55511e94517169a0d3","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-02-14 11:40:20.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"a1b93e298ac311646de2c7f5f2c821e95ee55d79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 11:41:00.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"9b6e5831d04a0a94aee3f9714d4843e068218b63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 12:42:26.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b69fa4e70d9847d488e3447f3fe2c35d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d34390520434a80a54987de5d8ecc99 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f577bc3ceb3423e9fc47e6e2c13f32b : SUCCESS in 9m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0535e4c1cb94023bb66680f010d49f9 : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d2e656081047c5949d5e96780241d6 : SUCCESS in 7m 22s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b6e5d72f8974cbb90482c698bb56ad3 : SUCCESS in 46m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3e3bb8be4c14bf3b1254f04339d2bb7 : SUCCESS in 43m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/862c377448784efababc53e784e8f50f : SUCCESS in 41m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/de3ebd3201cb4ae6805e3340571bc241 : SUCCESS in 49m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5f86ee37fb6748268004efcba1b9bd1b : SUCCESS in 42m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9a5d8e6efad147baaf1824f96ae0be67 : SUCCESS in 39m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1d53446c3139439a8c7cc4058211e0eb : FAILURE in 34m 23s","accounts_in_message":[],"_revision_number":13},{"id":"43418b9e03afc4ab38330be43f398885f84e760c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-02-16 07:53:42.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5657dcc9f99a99a20c4b5ac3de90768da1fd10ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 10:14:29.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/631cbeab4f2a439bb7c7fdcc457d7332\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fea890dea0041deae1ddaca1d1075d7 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/855e271397a6458194f8f4b1f3fae2a8 : SUCCESS in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e80b770a0864b03a22dbd661419a52f : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75fe2a0d3b34c7e985e97fc506e37e4 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/103708eb57a84f83ba811a3ee425fc54 : SUCCESS in 12m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/aebd80ad9c1644878137a13f08ddc580 : SUCCESS in 40m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee5b2dcb0d9342c2af5229207006a387 : SUCCESS in 49m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d3a7d240ca24fb5aa3233311b15547a : SUCCESS in 49m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e150c98a7f0247439be27cc31d410871 : FAILURE in 41m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ffc66009f3642759299d2f887b95050 : SUCCESS in 45m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c65aa0b6194f4834aee60c423293d719 : SUCCESS in 43m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fc00443068454f23aee5c2953a65df8d : SUCCESS in 35m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/91d6edf5d0dd4a0f994fb28c1ac2b1b2 : TIMED_OUT in 2h 04m 52s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"031010eafb4d60907bd5bc5e7a01b58184b7798f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-02-16 17:13:47.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7550d1d0d3e0300166b6efbc0128df6956132186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 19:26:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8804152129f14a84963d6e0e4e0df325\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbde512b4834a0a8d6e37810c1429a0 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a188374d32664110b65597cd0f44f02b : SUCCESS in 9m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de6d4148c41c49dba84e82365de7766b : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6141c618bb4c47a8cd07fee7bdef71 : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a7dc95ec9ec42cabbc21b39c96b771b : SUCCESS in 13m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c99cbc2f4cc142868d42b2c4f5ecd61e : SUCCESS in 42m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51d4427a424649d3986b8ba8be4fa123 : SUCCESS in 41m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7f7bf625b574a8e82e7273bf71cc378 : SUCCESS in 41m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fbe3e8665cc9445cbd0b0d58e7d71d93 : SUCCESS in 45m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5af1a9b03cb14604a2af22bf36dc197a : SUCCESS in 44m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f93c415fc42e449c868e33d53ecccdbb : SUCCESS in 43m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bb783677dbfa4d85abd512b15db36ee3 : SUCCESS in 41m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/096d33daaf014de79647216fc5aa0f10 : TIMED_OUT in 2h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1eb9b6d1084fdcc76a40be20df0139006a262961","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-05-17 08:28:59.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"775e2cbbd17bb98ae87b91ef3315dd5744f20c85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 08:29:34.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"4b83a7393a6bcf11930f62eb73f77c3fd45ac880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 09:20:39.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc1655186d54ad6b192690eb7c19d3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b8b8c097704c6c8b98e120719d61c3 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee27c48226eb4e46a480193ac43e87b5 : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2ca25bfa7e3473fae5eca106585f24e : SUCCESS in 10m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2574efe2644f446ba302ba313d040ae8 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91243d955f44a0886e90ee1d7576ee9 : SUCCESS in 7m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/846f80ad95694690876adeea54b14125 : SUCCESS in 42m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a3ffdf6e56441d78131c80495a87392 : SUCCESS in 38m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af5dc85640654107b2b01951824c891c : SUCCESS in 34m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e42c15e8f51a4a2e9c2ae0c40147aade : SUCCESS in 36m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/aae1eae0978b4658ba48659bfbbb2f3e : SUCCESS in 46m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cb383041f0a040ec9a5d0551253d7ee4 : SUCCESS in 40m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/be116aaf429e4923a9e0f8fa11f161d1 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":14},{"id":"3f7aa58b635d74bd3c7bbc7b0463e38bcbb67203","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 09:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"bb41116f57aebc0d15c28a0e70e873c7d8c6224c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 09:21:39.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd0f75ace3f4c7d8d4e7dfb4af563a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353f9fba153449fc8cdcf17f2c68e710 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"9f52e2b6f39306e79e9291e58985e376243aba63","revisions":{"78be33e32247309371ce1c1b27c9af273c8e482c":{"kind":"REWORK","_number":1,"created":"2023-01-04 11:14:39.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/1"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/78be33e32247309371ce1c1b27c9af273c8e482c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/78be33e32247309371ce1c1b27c9af273c8e482c"}]},"branch":"refs/heads/master"},"61bf326cc20659464d4f1c82b8aa34395a31652e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-04 11:49:42.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/2"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:49:42.000000000","tz":0},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61bf326cc20659464d4f1c82b8aa34395a31652e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61bf326cc20659464d4f1c82b8aa34395a31652e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4e17f2158fe401827a8f257b186f117326e03ab5":{"kind":"REWORK","_number":3,"created":"2023-01-04 13:01:17.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/3"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 13:00:30.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e17f2158fe401827a8f257b186f117326e03ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e17f2158fe401827a8f257b186f117326e03ab5"}]},"branch":"refs/heads/master"},"2bfab2de6aeee75916b9549cd54ae0eb1b958580":{"kind":"REWORK","_number":4,"created":"2023-01-04 13:11:48.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/4"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 13:10:59.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2bfab2de6aeee75916b9549cd54ae0eb1b958580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2bfab2de6aeee75916b9549cd54ae0eb1b958580"}]},"branch":"refs/heads/master"},"0d7ac7f35233ba95d46acc3b6007584486a8ee10":{"kind":"REWORK","_number":5,"created":"2023-01-05 10:34:28.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/5"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-05 10:20:22.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0d7ac7f35233ba95d46acc3b6007584486a8ee10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0d7ac7f35233ba95d46acc3b6007584486a8ee10"}]},"branch":"refs/heads/master"},"84e84cb5a7487134ae8f3c8213f8094ee455874e":{"kind":"REWORK","_number":6,"created":"2023-01-06 11:25:32.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/6"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-06 11:24:52.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84e84cb5a7487134ae8f3c8213f8094ee455874e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84e84cb5a7487134ae8f3c8213f8094ee455874e"}]},"branch":"refs/heads/master"},"e0e31c857cfbcf122a723bba901f1d4bd2c6cc44":{"kind":"REWORK","_number":7,"created":"2023-01-06 11:27:24.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/7"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-06 11:27:10.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0e31c857cfbcf122a723bba901f1d4bd2c6cc44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0e31c857cfbcf122a723bba901f1d4bd2c6cc44"}]},"branch":"refs/heads/master"},"5c717a816e8842b6c9697ea2fb6ba9f0bf14798f":{"kind":"REWORK","_number":8,"created":"2023-01-06 12:08:08.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/8"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-06 12:07:52.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c717a816e8842b6c9697ea2fb6ba9f0bf14798f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5c717a816e8842b6c9697ea2fb6ba9f0bf14798f"}]},"branch":"refs/heads/master"},"e3dd4e48cee9c33199f7cc29acf2392566a7938d":{"kind":"REWORK","_number":9,"created":"2023-01-06 13:04:11.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/9"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-06 13:03:46.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3dd4e48cee9c33199f7cc29acf2392566a7938d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3dd4e48cee9c33199f7cc29acf2392566a7938d"}]},"branch":"refs/heads/master"},"b79fa6c458b1df699bc0d231677ac38217a3996e":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-01-10 06:54:48.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/10"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-10 06:54:48.000000000","tz":0},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: https://blueprints.launchpad.net/mistral/+spec/distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b79fa6c458b1df699bc0d231677ac38217a3996e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b79fa6c458b1df699bc0d231677ac38217a3996e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"80e87a3335c43585f49befb2405a122e49d9abe0":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-01-11 07:39:28.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/11"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-11 07:39:28.000000000","tz":0},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: https://blueprints.launchpad.net/mistral/+spec/distribute-mistral-operations\n\nLink to failing Test case review: https://review.opendev.org/c/openstack/mistral-tempest-plugin/+/869652\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/80e87a3335c43585f49befb2405a122e49d9abe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/80e87a3335c43585f49befb2405a122e49d9abe0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d47a678387cd24e19d4c118bc1adfe7b5e03b24d":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-01-13 11:24:43.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/12"}}},"commit":{"parents":[{"commit":"b30ff6f95525dc1dbdba1642ced07e008f58df14","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b30ff6f95525dc1dbdba1642ced07e008f58df14"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-13 11:24:43.000000000","tz":0},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: https://blueprints.launchpad.net/mistral/+spec/distribute-mistral-operations\n\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d47a678387cd24e19d4c118bc1adfe7b5e03b24d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d47a678387cd24e19d4c118bc1adfe7b5e03b24d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4fd8b0ca84c5c5743fd31c8056be38fbd5c04c4d":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-01-19 05:52:44.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/13"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-19 05:52:44.000000000","tz":0},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: https://blueprints.launchpad.net/mistral/+spec/distribute-mistral-operations\n\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4fd8b0ca84c5c5743fd31c8056be38fbd5c04c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4fd8b0ca84c5c5743fd31c8056be38fbd5c04c4d"}]},"branch":"refs/heads/master","description":"Rebase"},"9f52e2b6f39306e79e9291e58985e376243aba63":{"kind":"REWORK","_number":14,"created":"2023-02-16 07:53:42.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/93/869193/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/869193/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/869193/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/869193/14"}}},"commit":{"parents":[{"commit":"4289317a914cd2542a23ccc2f3cdcb0d7fd00676","subject":"Merge \"Task skipping feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4289317a914cd2542a23ccc2f3cdcb0d7fd00676"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-01-04 11:14:15.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-02-16 07:49:42.000000000","tz":330},"subject":"Starting tasks via RPC","message":"Starting tasks via RPC\n\nAfter this patch mistral will run tasks using RPC which will distribute tasks amongst available engine threads. this will enhance performance in case of executing huge executions containing multiple tasks.\n\nImplements: blueprint distribute-mistral-operations\n\nChange-Id: I0b7202589eee68ba5560bf2aa60fbbd6118f3719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f52e2b6f39306e79e9291e58985e376243aba63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f52e2b6f39306e79e9291e58985e376243aba63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
