)]}'
{"id":"openstack%2Fmistral~870074","triplet_id":"openstack%2Fmistral~master~Idcb276912582bb097dc5c1c9692facde63d5886b","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-01-19 10:24:21.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-01-19 10:24:21.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-01-19 10:24:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idcb276912582bb097dc5c1c9692facde63d5886b","subject":"Fix sorting error when type is different","status":"MERGED","created":"2023-01-13 10:38:12.000000000","updated":"2023-01-19 10:24:21.000000000","submitted":"2023-01-19 10:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870074","meta_rev_id":"1b467d1aa889cfc926689d4c95feb52a4fc8f71b","_number":870074,"virtual_id_number":870074,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 10:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 08:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":1,"date":"2023-01-19 09:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-13 13:18:58.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-01-13 13:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-13 18:52:28.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2023-01-13 18:54:51.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"dff3ae1c29dab92671f0cdb431c756f503f66485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-13 10:38:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3732cfa954b4b24f0c5fae2b27146550dd7e3fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-13 11:31:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ace59df974c9bdfec3e87e499c07534fb83df41","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-01-13 13:18:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be1e6afccc350995adeef947ccf6739d73b92351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:42:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8ec4c0f0f304f59a15329b3fa0155b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d201e6c3fd4748abb724127f6dc630 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575eb850036f4a0ab411661868db68d8 : FAILURE in 8m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a003038643f548a8a0b921e91d98324b : FAILURE in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610c3b11fe2e427a9d1a0771720a0234 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0181bea59238404eb3c8112b53ba27ad : FAILURE in 11m 35s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/92fb78a7de984f48993f87f076188241 : SUCCESS in 40m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10af6bb770004f1782a0f3893ce1f854 : SUCCESS in 50m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e4ab017e8b6471f8f5379f9257c0b2f : SUCCESS in 33m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/40c47f2cf5024ab0ad88c0bebea380f0 : SUCCESS in 41m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6dc862c49c72432b9b4b951a62a07480 : SUCCESS in 47m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/78dcba80b2a344dd929ed9e9e02e5455 : FAILURE in 34m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5aa8f0712f4f4f2b9b08f01ed288dfd8 : FAILURE in 35m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b3f809496a3348c39849060d60e4d215 : TIMED_OUT in 2h 02m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2573afe2e6bd53ea820c596a71563432889d8936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-13 14:57:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31add343cb9149fbc04a54a7cba8eea2d404e012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 16:52:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d724a54e6f40ef9803f39e71a227fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572283a3cb8744a4a9332cf50b7af331 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54290d1ab2ab4ee09ba01399d1e9de3f : SUCCESS in 9m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf80ee037e2b4941938d913fdac9cf28 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecdcdeacbe34b7caabb062c51f703cd : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b074563a5f346ab82db28d7936490ac : SUCCESS in 12m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e707a8f2e9da41b1958b88edbefddbfd : SUCCESS in 52m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd8aa008610143f596ca0af0096852b0 : SUCCESS in 41m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/988fd41d1af94eb39308c53e9b405440 : SUCCESS in 49m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cf157cda202d48d4b96b8d34c25e695d : SUCCESS in 47m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2d3208d1990f40ae85bd060d6bffa146 : SUCCESS in 40m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/60724b74a39846aea16bf55a395f14b6 : SUCCESS in 40m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/15f35cbc034f448aa50efa3c023ad646 : SUCCESS in 32m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e02ac9b63eae4152bf7ecd07af9d6846 : FAILURE in 1h 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca3c733f0cb604efcac604ff68991226a6bf1090","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-13 18:52:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f36fc42b6ea237d91bef4969af398322a5930ce1","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-01-17 09:32:23.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"763bc555bab171db91734c902e82b560ea96356c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 09:33:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc6c35da3ca59214d0b04f99fb9935d151999634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:28:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc6af3688bf34b39b67107da5b68657b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef8c817f14a456ca9ec2ebc6c9b013a : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57dcc93dbc9e4df2943c7d70205bac26 : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02e02fb6de4247e19e15b4acb2b295a8 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6b2d7dd5494bf38d20c39cb14ee5a3 : FAILURE in 6m 08s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cdbbd9294bbb4862bb5fe8bd8f4fe514 : SUCCESS in 50m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe3d9f9f179545dba7d1d05a8c7f327e : SUCCESS in 37m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02b7cc7621454a6c99f7f5467aae20e3 : SUCCESS in 31m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9943c19eb4454523b1e235dd834022ed : SUCCESS in 47m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2cf441626b0d4aff989bcd6ccda645b4 : SUCCESS in 39m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ae564dbcfa10442389e91f4771e70b5c : SUCCESS in 38m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/273d3769b9eb49db8272e42f4d9ffe75 : FAILURE in 33m 01s","accounts_in_message":[],"_revision_number":3},{"id":"fd6ee9327938a87cafa78292d7d49480ad06ab22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-17 16:33:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4cf2d0cdd9c7e4a649084fe0629e9bd75033ef58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 18:13:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29ce6b17336644bdabf82dcecd20f68a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1173d696a7154b09be2285dcfe0599a9 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8180b05bd44f0d8d18777d752da418 : FAILURE in 8m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/698a9c829a0b4714938d08ea225ba341 : FAILURE in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5a6f43acc345448e069c28dbe8ebeb : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/549b5c562360495c9b4841bb0a78fbc5 : FAILURE in 10m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cc01e794ec1e4caaaa34385ef6ae9391 : SUCCESS in 32m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da085284a1794085a365128799eef067 : SUCCESS in 33m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2ba0ea7d1f04a0a83eeed6aaff5dfd9 : SUCCESS in 33m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9161bd6875084cc883dad102de681330 : SUCCESS in 28m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f641409fd7dd42329f173d084773bdac : SUCCESS in 34m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d3434ef084334f49ad56d7225ee0590e : FAILURE in 34m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0b1397dbbf8947248cba1c87a47968bf : FAILURE in 33m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a788cdb3f6c7417f8061cd13810cfec0 : FAILURE in 1h 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e7b9b6c6cbc3a24e0bd7906f24e249dc0ac6176b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 17:02:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c791a522a0131b39a9b044b992b025756358714c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:54:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76a930afb33e48e98e3be5f8725a43f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dac3b3281543bd828775b4b2a01e7d : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c74a10dd624f413891624fbfab89521e : FAILURE in 9m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6861ee2b82b34dc29ec674af51b54226 : FAILURE in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fddd6106c54f78aa1f86f590410651 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b131f19e2c724995b8a3e306afc8dcb4 : FAILURE in 12m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0dc6817855ce4f67867d47dc6311c5b7 : SUCCESS in 43m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/003ebf5408964ef7b347cca7c446c183 : SUCCESS in 38m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c1bac9eba3144f9839942759d78cc51 : SUCCESS in 40m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a4f3f348e66a4e2ca93992067d414659 : SUCCESS in 48m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a9f324dc7b044b19b17e95e8e6587df6 : SUCCESS in 50m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d64aa3dedf9b49d7bc54d2aea971ac3c : FAILURE in 36m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/45444424ea7d430cbdb441b6e9e73d6f : FAILURE in 32m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f7ce10e5c8c748af9527199bdb3435e0 : FAILURE in 1h 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8eda95e8f67becae6d4015e185c18a3ca4df9229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 20:19:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"36a5c14e089d256aaa2d2dbe83c990398e9c6540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 20:20:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"850f0f92051403dbba1655e9f0c5ed1d17bcd08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 22:41:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991b7f3f7d2d487ca3b97473ce09e203\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf4b7abdfbb49d6bcfd5fb9ded0ff10 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/723b7378af384a869d74eb2ca6a11afd : SUCCESS in 9m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bef44948c734ab3bb45c7dd9efe9522 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255799c09caa4f478366ea7de8b8d5c7 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f7e688ee3bd4977919c317ebbc26fee : SUCCESS in 12m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7fa38190cf4241b1bbf00ff5cf17f308 : SUCCESS in 39m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4704d514d94449faa363e67a7334dfd : SUCCESS in 43m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7dc02df4182f48ef9c3a9d7d2c0a5de1 : SUCCESS in 49m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/30dc54b2c2534b2abd870e6602d15bc5 : SUCCESS in 40m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b0d3d90e8c3c484390858f3b38386861 : SUCCESS in 34m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f5973ff7e3124e0a952286d5444fe429 : SUCCESS in 37m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a0ea07c91f784b27a0bbdbaca4f189dd : SUCCESS in 34m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4a69d9faeecc4c6392b01e7ae9e6caa0 : TIMED_OUT in 2h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cd1342e42e32d868a30b9c5fc88420a472eb0e74","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-19 08:07:43.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"852e3d39e45e2d45abeb8ccd2c65da0d6de53a90","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-01-19 09:16:41.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f230b038ec2961d0b4903cc5a0470817009570c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 09:17:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"76e252540f535f8865017fd266d9460a5b27838a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:24:21.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d4a33655f5e47d283272f69952151ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c81e833fc84ca6920d8be1b6ca4283 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91ea13125f3649f2be5ae13ce6625ba9 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d41f91ab54f477d971155133bd50fd3 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7ea582cbaf459d95f31cab5ebb8e80 : SUCCESS in 7m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a6ced178ad34444d8f66fbdc7209456f : SUCCESS in 33m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc3c32f04b7648778429b0baa12f711b : SUCCESS in 47m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb7dbd2e52ef4954b86f7bd7fb28ae24 : SUCCESS in 45m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/053dd98ff3234a02952c3ada271b14cc : SUCCESS in 44m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f5f64c73c132428cbaa77f2ebb4655af : SUCCESS in 32m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0121bfe370b94c8193b04e5d31505fe8 : SUCCESS in 39m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/406b8f4d79fc4e52a08d3b200ca72ec0 : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":7},{"id":"1b467d1aa889cfc926689d4c95feb52a4fc8f71b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:24:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b6a47992e7299dbb4ed684ecaf05d35d669c38b8","revisions":{"a99e6e1dfb00587ed5104dfa96151a00c349cd12":{"kind":"REWORK","_number":1,"created":"2023-01-13 10:38:12.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/74/870074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/1"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a99e6e1dfb00587ed5104dfa96151a00c349cd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a99e6e1dfb00587ed5104dfa96151a00c349cd12"}]},"branch":"refs/heads/master"},"ad78db14c55f7caeaa555a0e7e4d44800cb62e4a":{"kind":"REWORK","_number":2,"created":"2023-01-13 11:31:48.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/74/870074/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/2"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 11:31:38.000000000","tz":0},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad78db14c55f7caeaa555a0e7e4d44800cb62e4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad78db14c55f7caeaa555a0e7e4d44800cb62e4a"}]},"branch":"refs/heads/master"},"28a1a71eb339100e81ba5e50ba337067e186b6dd":{"kind":"REWORK","_number":3,"created":"2023-01-13 14:57:45.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/74/870074/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/3"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 14:57:39.000000000","tz":0},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28a1a71eb339100e81ba5e50ba337067e186b6dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28a1a71eb339100e81ba5e50ba337067e186b6dd"}]},"branch":"refs/heads/master"},"c29c81cda98db45479b617bdeb78b4a8220d715e":{"kind":"REWORK","_number":4,"created":"2023-01-17 16:33:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/74/870074/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/4"}}},"commit":{"parents":[{"commit":"d5caed1698e2bdd5aac9c702c2bd9b0d35878326","subject":"Force dnspython\u003d\u003d\u003d2.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5caed1698e2bdd5aac9c702c2bd9b0d35878326"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 16:32:58.000000000","tz":60},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c29c81cda98db45479b617bdeb78b4a8220d715e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c29c81cda98db45479b617bdeb78b4a8220d715e"}]},"branch":"refs/heads/master"},"6e562880fe8c9d0b9a6481b1c3d9751bf9456c71":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-18 17:02:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/74/870074/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/5"}}},"commit":{"parents":[{"commit":"c77a51a05a9b0865179f4f597d9a72a8faa6c17c","subject":"Fix doc builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77a51a05a9b0865179f4f597d9a72a8faa6c17c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 17:02:03.000000000","tz":60},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e562880fe8c9d0b9a6481b1c3d9751bf9456c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e562880fe8c9d0b9a6481b1c3d9751bf9456c71"}]},"branch":"refs/heads/master"},"82d67a65726e2e5bee4fc365cbae68217fceb6a5":{"kind":"REWORK","_number":6,"created":"2023-01-18 20:19:11.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/74/870074/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/6"}}},"commit":{"parents":[{"commit":"c77a51a05a9b0865179f4f597d9a72a8faa6c17c","subject":"Fix doc builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77a51a05a9b0865179f4f597d9a72a8faa6c17c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 20:18:34.000000000","tz":60},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82d67a65726e2e5bee4fc365cbae68217fceb6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82d67a65726e2e5bee4fc365cbae68217fceb6a5"}]},"branch":"refs/heads/master"},"b6a47992e7299dbb4ed684ecaf05d35d669c38b8":{"kind":"REWORK","_number":7,"created":"2023-01-18 20:20:34.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/74/870074/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/74/870074/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/74/870074/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/74/870074/7"}}},"commit":{"parents":[{"commit":"c77a51a05a9b0865179f4f597d9a72a8faa6c17c","subject":"Fix doc builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77a51a05a9b0865179f4f597d9a72a8faa6c17c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-13 10:36:32.000000000","tz":0},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 20:20:28.000000000","tz":60},"subject":"Fix sorting error when type is different","message":"Fix sorting error when type is different\n\nThere was a logical flaw where we compared types\nthat were different causing a TypeError.\n\nThere was also a flaw in that we default to the\nname of the action when sorting if the key does\nnot exist, to compare the data both value should\ncome from the same key as well otherwise they\nshould be treated as not equal since the data\ncannot be.\n\nChange-Id: Idcb276912582bb097dc5c1c9692facde63d5886b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6a47992e7299dbb4ed684ecaf05d35d669c38b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b6a47992e7299dbb4ed684ecaf05d35d669c38b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}}]}],"submit_requirements":[]}
