)]}'
{"id":"openstack%2Fmistral~870840","triplet_id":"openstack%2Fmistral~master~I5f86f8264792c8e651c0bb2d2f63151657fea1be","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-01-19 10:24:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f86f8264792c8e651c0bb2d2f63151657fea1be","subject":"Fix doc builds","status":"MERGED","created":"2023-01-17 16:33:13.000000000","updated":"2023-01-19 10:25:32.000000000","submitted":"2023-01-19 10:24:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870840","meta_rev_id":"7c3b52b0925b6f53ef4bc025823772c90e55bb6e","_number":870840,"virtual_id_number":870840,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 10:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 08:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":1,"date":"2023-01-19 09:16:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 18:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 09:16:50.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"4645690a59ae9f0223d99f3e860a47e8a662b2c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-17 16:33:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ac6c95ebc1b5e65ed15a5e75cc8feabdaee2c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 18:56:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d709c98b6b4c4a46b12c85ef1a1da1db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff8592e49c6543d0885e06bbd799575e : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65845142ba444a62922339d1d88bbf6a : SUCCESS in 10m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6b9eea1bc864544bacd5aa6348b8e80 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0773205701c477489b9459a007f0148 : SUCCESS in 8m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea9a25dd333d4743865d831a5af12021 : FAILURE in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4238d98f71f146baa8e23b07e06a4e32 : SUCCESS in 15m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d8bafe1100a346c994aed77d843c1ddf : SUCCESS in 43m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/872faa440fd643dea91d6ed51045337f : SUCCESS in 43m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a248bf1f7474df682d3ca59d25b4df8 : SUCCESS in 47m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bc8927d9bdff4015a0a241bfbbba175e : SUCCESS in 1h 04m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3ba3e6abe2b043a8a17a7363a0ed1217 : SUCCESS in 49m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1ef5ba7815d74dc393705f37a28e156f : SUCCESS in 44m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/02de8fdbfe7b4a3a8576befb81f5f4b3 : SUCCESS in 39m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9123480965fa4fa79d80dc258b118dbf : TIMED_OUT in 2h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd8cd966696a81b5f032a2a70b40ad97011b576b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-01-18 16:30:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7cbcec16867f4681682189293f2ec4514463436b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 17:02:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c1cfee7e5cf355b96119082579cd625a88d0d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:15:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4ea14e6fb44b1d91e352264b348547\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f25adbd94e4fa5809dd52dd30179e2 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ecdc63cc28d41c29b1b8c6d5525109a : SUCCESS in 9m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78663b5358504c3cbf8e2d9740d97dec : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e09e9ca599435ab3b21ded00ff0d16 : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dc41fb7befed4d09b3f40f3bc5b479de : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28b493301e704fdda0b328162b4da798 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e7907ec30664baaa54461e993b8e343 : SUCCESS in 13m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/86847e98a9d94d3cb5a293e54f2b5e74 : SUCCESS in 46m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1fec680b3f842c1addcf88ead2bfc50 : SUCCESS in 45m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58fc786db08645459bc4ca4dfa0a3d63 : SUCCESS in 45m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b78e138e303d463abc66aaf15bd47a56 : SUCCESS in 48m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/107dba1d5d5247a9b64d0409e3768e52 : SUCCESS in 36m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6636b493852b49ab880666c6c0cd585d : SUCCESS in 42m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dabe3d3c24084bd5a37846a084646cdb : SUCCESS in 37m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5e3ae45166a0443eb82b699693e198a9 : TIMED_OUT in 2h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dcbe6590687d6e0d49794537f02fed13803e84d3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-19 08:07:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f5ee3bac89eefdd6c1391e2de7e5cea7ca48d2bc","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-01-19 09:16:50.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a579b7afc21e82206338f9602026d35f36c5aaa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 09:17:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8f0aae5217a9d6545b7f15e564c289e1f808dcc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:24:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8361669d387c4aa7acfd1777da388d2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a7d1ff00964f30baf58cac02e4dbf5 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d006b23bb23e4b83a614a7e0d49e43a3 : SUCCESS in 10m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd4fc21b92994d598d25bbd4abb2fcb8 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ed3f8a242f40c88d748c8a25ba5a87 : SUCCESS in 7m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9811cd8147a444383ae44876e79ebd4 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a7d7219d3094b5688409628ea449cf0 : SUCCESS in 3m 54s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8c8616abba1c4fb6bc179bf4febfd55c : SUCCESS in 35m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ad63702f8c84ccaa75ba58188ab46d3 : SUCCESS in 57m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54ff225f88af4d98b5b544fc7997c77a : SUCCESS in 51m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7b02260689684dd5b5c041cf8b5137d5 : SUCCESS in 46m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b600e0dee4434ac79f0aeff8c6cba2f1 : SUCCESS in 48m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/80f8f55eca8e4e27b5125a389ec39582 : SUCCESS in 38m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/069c028c818c457283f99f73880d95de : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":3},{"id":"9bc9bbb22dc347b36d6a75dd3213e7657d169adf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:24:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7c3b52b0925b6f53ef4bc025823772c90e55bb6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:25:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e79f3fd44640c0b932882ac9d2b950\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5f2481c5da43e685d34e9cbafad475 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cbf7d57b2284c508b75495fb4f745c6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c77a51a05a9b0865179f4f597d9a72a8faa6c17c","revisions":{"d5caed1698e2bdd5aac9c702c2bd9b0d35878326":{"kind":"REWORK","_number":1,"created":"2023-01-17 16:33:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/40/870840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/870840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/870840/1"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 16:23:25.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 16:32:58.000000000","tz":60},"subject":"Force dnspython\u003d\u003d\u003d2.2.1","message":"Force dnspython\u003d\u003d\u003d2.2.1\n\nA new release of dnspython has been release, which is breaking the docs\ntests. See [1]\n\n[1] https://github.com/eventlet/eventlet/issues/781\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I5f86f8264792c8e651c0bb2d2f63151657fea1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5caed1698e2bdd5aac9c702c2bd9b0d35878326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5caed1698e2bdd5aac9c702c2bd9b0d35878326"}]},"branch":"refs/heads/master"},"58be2ebb644ff4f781c0ed0b25f3069721d8f0d8":{"kind":"REWORK","_number":2,"created":"2023-01-18 16:30:18.000000000","uploader":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"ref":"refs/changes/40/870840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/870840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/870840/2"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 16:23:25.000000000","tz":60},"committer":{"name":"Axel Vanzaghi","email":"axel.vanzaghi@axellink.fr","date":"2023-01-18 16:29:54.000000000","tz":0},"subject":"Force dnspython\u003d\u003d\u003d2.2.1","message":"Force dnspython\u003d\u003d\u003d2.2.1\n\nA new release of dnspython has been release, which is breaking the docs\ntests. See [1]\n\n[1] https://github.com/eventlet/eventlet/issues/781\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I5f86f8264792c8e651c0bb2d2f63151657fea1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58be2ebb644ff4f781c0ed0b25f3069721d8f0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/58be2ebb644ff4f781c0ed0b25f3069721d8f0d8"}]},"branch":"refs/heads/master"},"c77a51a05a9b0865179f4f597d9a72a8faa6c17c":{"kind":"REWORK","_number":3,"created":"2023-01-18 17:02:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/40/870840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/870840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/870840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/870840/3"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 16:23:25.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 17:02:03.000000000","tz":60},"subject":"Fix doc builds","message":"Fix doc builds\n\nA new release of dnspython has been release, which is breaking the docs\ntests. See [1]\n\ndnspython is fixed to \u003c2.3.0 in upper-constraints, so we should rely on\nrequirements.txt from mistral to trigger the correct build.\n\n[1] https://github.com/eventlet/eventlet/issues/781\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I5f86f8264792c8e651c0bb2d2f63151657fea1be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77a51a05a9b0865179f4f597d9a72a8faa6c17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c77a51a05a9b0865179f4f597d9a72a8faa6c17c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}}]}],"submit_requirements":[]}
