)]}'
{"id":"openstack%2Fmistral~870869","triplet_id":"openstack%2Fmistral~master~I72ce8851ea5b379a8af75b32fb600691638836af","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-01-19 10:26:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72ce8851ea5b379a8af75b32fb600691638836af","subject":"Remove tenant when building mistral context","status":"MERGED","created":"2023-01-17 21:49:09.000000000","updated":"2023-01-19 10:27:20.000000000","submitted":"2023-01-19 10:26:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870869","meta_rev_id":"24b30b75bd54f1c6d7f705835084075846ce6ee9","_number":870869,"virtual_id_number":870869,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 10:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 08:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":1,"date":"2023-01-19 09:16:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 23:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 09:16:22.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"7a95e652fc4c7dc450f9beeed9e2418936eadfc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-17 21:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0888e403c095710590cc1cc363fda43e70718959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 23:59:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/023de72457cd4c09ae7677cfcf717d2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94dc572a04834c0383a734de19e5f70b : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cdb48fc3d7644d8aa6c18cd5a861752 : FAILURE in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1821560083394acd94cc3db0e287805d : FAILURE in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b467897a6dd4657882060e8ee0b1164 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/918bc2299acd482eb5668b860c3fe173 : FAILURE in 12m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ec072e3bb6d44cc89539548a012d7f59 : SUCCESS in 42m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1487556c5bd4c3aa7b39e7da42d2ab5 : SUCCESS in 43m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5acaf374ee984991a2093485e5218e3a : SUCCESS in 43m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/77c51f47e0a9449aa8d93d098f718878 : SUCCESS in 45m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/54c4a1d6cb70461c9804b8649c323a59 : SUCCESS in 40m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/06985f53d3944d218cae46e5ffb8e498 : FAILURE in 40m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/003616ad28874be68d6a3d527ce7dc81 : FAILURE in 35m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1268320600dc4daeb1a9fa21ca533b63 : TIMED_OUT in 2h 02m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfaaa10e10d027e5dd80246fd983fb577de07b4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 17:02:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"34f5ff0e7bb12526a40bfdbc446ded6d238722f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:53:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f99a9eb4660407abbe2e0a512284d10\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540f5ba0c0434404982dca138e1f4abc : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/170a46af1cfc48c49170714dd53290fa : FAILURE in 10m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f9ba06fc38b473c942724e80281cdd8 : FAILURE in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca4a4b5e1ec4e12a63e0bbd17faf80c : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d1b05f05be94e03a9495319b8caace0 : FAILURE in 13m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/62e2cc88ac3745c898cbf1a838e06801 : SUCCESS in 38m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f45828306994eb18810c4630df59e70 : SUCCESS in 39m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e73c1e54f306476bb8b49fa872278b7c : SUCCESS in 42m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/38b3157b2da24733b95b370a02dc656a : SUCCESS in 43m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fcbc52f904bf4ea0be47cf068f2af903 : SUCCESS in 43m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/71db074664064699992f5d3921295847 : FAILURE in 38m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c10a92bf356046758b391ccdb0dbe099 : FAILURE in 34m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c0ede7cea6d148ed93f80a90f8d03203 : FAILURE in 1h 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"959ad739daf03127ee988966bc9be14afb75a0d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 20:19:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1b0723e250d0db229424137ed996a7a81358ec9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-18 20:20:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c6cca21f280fc8db420a7d7d7bbbe0968c8924c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:56:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfa4afa668af4b1397b78c29e1f8c257\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11837c7e204c49ad8211b231b52ac969 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2723ac05cf72478dab409a25d207cc6c : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bae12a234e942198d6b580437791465 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7698d73aa63947d593aad5f308eea50e : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5932d004114aa9b833f5ff701e24b9 : SUCCESS in 11m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/37803d2bde324aa5a4b5c6e6301f2330 : SUCCESS in 41m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb3a8fd4b4884f69b0cb0ac10a29292b : SUCCESS in 51m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea77774583224d69a0a72a6e3313d569 : SUCCESS in 44m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ea8ebd95c89c4ae2930c4fa4d0b50a62 : SUCCESS in 49m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/233037e76fc24be98877a0ce6095304d : SUCCESS in 47m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fbfc1b3b317148c1972ba4c3a7c476e2 : SUCCESS in 37m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/22c02d19b86f48f796459d18f021bc92 : SUCCESS in 39m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/238f48e1f7e940c9bb68db8f6d111236 : FAILURE in 1h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0bae5f8b469b451138aee96e9c0e3e50f17a2a4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-19 08:07:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7280bcb3f124d810034c95ad6824980b6c6c36cf","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-01-19 09:16:22.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2ee0710c974d50baed5d31e299f5f068b06ef5c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 09:17:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"00f6fd6a823daeb44d4280b79e54f69e1665d648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:26:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45104301a3944ac2a7fd1f2cefd135cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8caa396438fa4e6183827295978abe92 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e782d8808bc419b994f572ebfb22ee4 : SUCCESS in 10m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/995f67a23fd749fdaf447832912afc84 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349901c023814509913fe287e40a60bc : SUCCESS in 7m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c8101429ebc64e129d6f601d470fea49 : SUCCESS in 39m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8052440d6b2645579c6f6f7022a4b0ca : SUCCESS in 51m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8a8d05dd77942f0975c71e12d8de287 : SUCCESS in 43m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c9fb7f4da0840dfa65dd318a01a04db : SUCCESS in 41m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d5bf563cf0454ade9ebb797b4989aecb : SUCCESS in 44m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/72ce88b21a1e4318b278f8282692f516 : SUCCESS in 39m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a2a02c69744341758d0d6e6403af6e41 : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":4},{"id":"53f4b1b5eacb8d5cb300deb526959e7fcfc46e54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:26:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"24b30b75bd54f1c6d7f705835084075846ce6ee9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:27:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32123e2a57c34ee4b827ce9e218f9491\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78bbce94eb24073a99831d06b1921e2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"70af40becc8a90b83adcb93342c607ffe5add428","revisions":{"d7a5e6cad79e54eaad86cfb090ce80e28820c9ea":{"kind":"REWORK","_number":1,"created":"2023-01-17 21:49:09.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/870869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/870869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/870869/1"}}},"commit":{"parents":[{"commit":"dfa55abd9bf3404a775b1e711d689feb7a82893b","subject":"Refactor mistral documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dfa55abd9bf3404a775b1e711d689feb7a82893b"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 21:41:24.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 21:41:24.000000000","tz":60},"subject":"Remove tenant when building mistral context","message":"Remove tenant when building mistral context\n\nWhen building mistral context (using oslo.context) from environment, the\ntenant extra key was given, which is not accepted by oslo.context.\n\nThis issue was detected when debugging a mistral-client test:\nmistralclient.tests.functional.cli.v2.test_cli_v2.NegativeCLITests.test_target_action_execution\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I72ce8851ea5b379a8af75b32fb600691638836af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d7a5e6cad79e54eaad86cfb090ce80e28820c9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d7a5e6cad79e54eaad86cfb090ce80e28820c9ea"}]},"branch":"refs/heads/master"},"c3ac303a654a0b889c3014c6cc230ab83d8a5739":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-18 17:02:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/870869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/870869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/870869/2"}}},"commit":{"parents":[{"commit":"f1caf7f31098491b3c95887eb56db8af13b02c51","subject":"Refactor mistral documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f1caf7f31098491b3c95887eb56db8af13b02c51"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 21:41:24.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 17:02:03.000000000","tz":60},"subject":"Remove tenant when building mistral context","message":"Remove tenant when building mistral context\n\nWhen building mistral context (using oslo.context) from environment, the\ntenant extra key was given, which is not accepted by oslo.context.\n\nThis issue was detected when debugging a mistral-client test:\nmistralclient.tests.functional.cli.v2.test_cli_v2.NegativeCLITests.test_target_action_execution\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I72ce8851ea5b379a8af75b32fb600691638836af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3ac303a654a0b889c3014c6cc230ab83d8a5739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3ac303a654a0b889c3014c6cc230ab83d8a5739"}]},"branch":"refs/heads/master"},"761c2b51786e596ede033ada6b1a16cbb4df9bd0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-18 20:19:11.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/870869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/870869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/870869/3"}}},"commit":{"parents":[{"commit":"37fc8ca9feb4df6142c58b9e36c57465d6744f67","subject":"Refactor mistral documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/37fc8ca9feb4df6142c58b9e36c57465d6744f67"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 21:41:24.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 20:18:34.000000000","tz":60},"subject":"Remove tenant when building mistral context","message":"Remove tenant when building mistral context\n\nWhen building mistral context (using oslo.context) from environment, the\ntenant extra key was given, which is not accepted by oslo.context.\n\nThis issue was detected when debugging a mistral-client test:\nmistralclient.tests.functional.cli.v2.test_cli_v2.NegativeCLITests.test_target_action_execution\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I72ce8851ea5b379a8af75b32fb600691638836af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/761c2b51786e596ede033ada6b1a16cbb4df9bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/761c2b51786e596ede033ada6b1a16cbb4df9bd0"}]},"branch":"refs/heads/master"},"70af40becc8a90b83adcb93342c607ffe5add428":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-01-18 20:20:34.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/870869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/870869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/870869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/870869/4"}}},"commit":{"parents":[{"commit":"e75b6b356f4be44a24bc24ccfa4a3f8eca818b9a","subject":"Refactor mistral documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e75b6b356f4be44a24bc24ccfa4a3f8eca818b9a"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-17 21:41:24.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-18 20:20:28.000000000","tz":60},"subject":"Remove tenant when building mistral context","message":"Remove tenant when building mistral context\n\nWhen building mistral context (using oslo.context) from environment, the\ntenant extra key was given, which is not accepted by oslo.context.\n\nThis issue was detected when debugging a mistral-client test:\nmistralclient.tests.functional.cli.v2.test_cli_v2.NegativeCLITests.test_target_action_execution\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I72ce8851ea5b379a8af75b32fb600691638836af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/70af40becc8a90b83adcb93342c607ffe5add428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/70af40becc8a90b83adcb93342c607ffe5add428"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}}]}],"submit_requirements":[]}
