)]}'
{"id":"openstack%2Fmistral~876548","triplet_id":"openstack%2Fmistral~stable%2F2023.1~I0cf7e7a3eefa866076b585ec875b209713375d6b","project":"openstack/mistral","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I0cf7e7a3eefa866076b585ec875b209713375d6b","subject":"Update .gitreview for stable/2023.1","status":"MERGED","created":"2023-03-06 09:21:44.000000000","updated":"2023-03-07 11:08:16.000000000","submitted":"2023-03-07 11:07:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876548-create-2023.1","meta_rev_id":"f1ee153b60e5f7eb22fff31febab5cf0ae815dc8","_number":876548,"virtual_id_number":876548,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 11:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-03-07 10:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-03-07 10:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 10:13:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 10:09:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"88ace01626e32bceb4e044a1ad83fc1f187105c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-03-06 09:21:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11122b5b94d735d620cdd757a9c6155b3531e2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 10:13:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13ad3335be548dd84deaaa20529c030\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697c8e5711924deca01c5946ae343215 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfe23653935845368f2c1300f969291b : SUCCESS in 9m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9035009ce6bd4ba89544e83fa53a0f68 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ed577929a1483e84a4e476e045e61e : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819ca0b3daf84c0eae107bd4a7b46407 : SUCCESS in 13m 08s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9a17aecec8db49dd92ff68ec661f8b89 : SUCCESS in 40m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8e4894bb4dd74b098fedfe88852340d1 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6ec80b59cb23bffed5621d33914c7d87568b7dba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-07 10:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b01db4346fe1b53820c723ae41eb17676d3fa402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:10:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"821ff6d215439ea354b7d598dbbbb52f93bbb2bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:07:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56327d6ac83342b08d35484fd63afb77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92dd9631656e4d25a974c5404707975c : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60336d501c7048d897bb8d4ebeb7afd3 : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4804ab490704c3cafb7f03f47b7f982 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062054fb22564db7bc1cb36af1ef035f : SUCCESS in 6m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1b73e3e7b4ed4ca88b53562c8e5a6c10 : SUCCESS in 44m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6371758eab144cb587cae988294b2b82 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9b38d58b507daab4e2d7205f01b6e0e1756a8ae8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:07:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1ee153b60e5f7eb22fff31febab5cf0ae815dc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:08:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75910fd31d1d43fdabe6a778b2ac56b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edfd3d7e01245b7bb0d359e37328ad5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6198954ab0ff549f8864b97ac1a15f0285cf73de","revisions":{"6198954ab0ff549f8864b97ac1a15f0285cf73de":{"kind":"REWORK","_number":1,"created":"2023-03-06 09:21:44.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/48/876548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/48/876548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/876548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/876548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/48/876548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/48/876548/1"}}},"commit":{"parents":[{"commit":"4289317a914cd2542a23ccc2f3cdcb0d7fd00676","subject":"Merge \"Task skipping feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4289317a914cd2542a23ccc2f3cdcb0d7fd00676"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-06 09:21:43.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-06 09:21:43.000000000","tz":0},"subject":"Update .gitreview for stable/2023.1","message":"Update .gitreview for stable/2023.1\n\nChange-Id: I0cf7e7a3eefa866076b585ec875b209713375d6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6198954ab0ff549f8864b97ac1a15f0285cf73de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6198954ab0ff549f8864b97ac1a15f0285cf73de"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[]}
