)]}'
{"id":"openstack%2Fmistral~877932","triplet_id":"openstack%2Fmistral~master~Ie073d9ac8fd8580f7442370291814d99aad92066","project":"openstack/mistral","branch":"master","topic":"bug/2012246","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-20 12:17:00.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ie073d9ac8fd8580f7442370291814d99aad92066","subject":"[coordination] backend_url should be secret","status":"MERGED","created":"2023-03-20 09:31:18.000000000","updated":"2023-03-20 12:51:35.000000000","submitted":"2023-03-20 12:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877932-bug/2012246","meta_rev_id":"007afb4309e96f97fa83561b42014e69304ab216","_number":877932,"virtual_id_number":877932,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 12:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-03-20 09:44:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-20 11:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-20 11:55:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 09:44:09.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-03-20 11:50:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 11:55:29.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"64cd4a01fec11cb2619adaf2fe78057343d10b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-20 09:31:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e9f930d6ce015a7de72af4c158704a08dabbb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-20 09:39:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"94a127da6ac1636be5149931afd0b688834a0edb","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-03-20 09:44:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"efb266456651df57bbe9ff82364717b1e75c9b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:50:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ed3b278997428f815bfb640d46830f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0440b9d067d0463789c956a7b3ec8dd1 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a55af9eacf428ca429f966be7ebf18 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecae6c1cbf9c4156bf5c9efdfbc34af9 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81f74db82bb471392985a06399306b2 : SUCCESS in 7m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81ad347fe5ce40339becd14f111f7dff : SUCCESS in 12m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f64639c749bf487085fd1bd811ddfb06 : SUCCESS in 51m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e26bc5ed44cb46908fe5e710194b7f23 : SUCCESS in 43m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bde184c5eb5402688573ef4c6ce1100 : SUCCESS in 1h 11m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3cda997b28ac4387bd2f0b81063a5d60 : SUCCESS in 45m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6670ac59edc34a5590c659d4ffd259fb : SUCCESS in 51m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4ce4122d716f49509b5c60331a4c2110 : SUCCESS in 46m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c10d31b2f5b540c0ba16ad3a9fec91a1 : SUCCESS in 40m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9eb392b97b73472bac81473f2bec77c8 : TIMED_OUT in 2h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9204f16d6f61c0fe25f57674de341412db811f39","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-03-20 11:55:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7554e82edef4c076cf4827a96f03072f6e587afd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:56:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"02cbd3e45c7e5804a56c00187b4563fc35cbc112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 12:50:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac44bc75493f40d28adbc47e542515a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe96885e805940b1b5f3714e241f6f32 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0c5b4ad9984611a93d3de7b2862c29 : SUCCESS in 9m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05f59555c43f450fb9d677c7dea05c16 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1b638135294e628c3cbc27ffe0aa98 : SUCCESS in 7m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/99827fe1755449b1b539e28feb3fde78 : SUCCESS in 36m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cd6c5d1cef741dcaa8507bc9a9ee6e6 : SUCCESS in 36m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92dc1812f2264f4998d93ec76cc8baaf : SUCCESS in 48m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d4225a7ed88e4d6a83ba830669604357 : SUCCESS in 49m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9f1e903e34a240f4aa50d51dc521d398 : SUCCESS in 45m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a23811825be645458d367e12e23332ab : SUCCESS in 41m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2de204a9e7854baa813416c7ac1b0c7c : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"41cc222fa0e7d0ead91aaefae2fbe2071c42ec0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 12:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"007afb4309e96f97fa83561b42014e69304ab216","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 12:51:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e0419710ec42c2826c56f098bfcbb6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf558b4a0994d13a095dc3bca2190b0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f9f1f3ffb9db5fb6d32d3f05b21a21d51257e0e","revisions":{"9382572ee1441446079df9c6427037078e7ba8eb":{"kind":"REWORK","_number":1,"created":"2023-03-20 09:31:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/877932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/877932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/877932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/877932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/877932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/877932/1"}}},"commit":{"parents":[{"commit":"4289317a914cd2542a23ccc2f3cdcb0d7fd00676","subject":"Merge \"Task skipping feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4289317a914cd2542a23ccc2f3cdcb0d7fd00676"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:30:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:31:15.000000000","tz":540},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain a password string.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nChange-Id: Ie073d9ac8fd8580f7442370291814d99aad92066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9382572ee1441446079df9c6427037078e7ba8eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9382572ee1441446079df9c6427037078e7ba8eb"}]},"branch":"refs/heads/master"},"1f9f1f3ffb9db5fb6d32d3f05b21a21d51257e0e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-20 09:39:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/877932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/877932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/877932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/877932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/877932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/877932/2"}}},"commit":{"parents":[{"commit":"4289317a914cd2542a23ccc2f3cdcb0d7fd00676","subject":"Merge \"Task skipping feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4289317a914cd2542a23ccc2f3cdcb0d7fd00676"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:30:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:39:45.000000000","tz":540},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain secrets.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nChange-Id: Ie073d9ac8fd8580f7442370291814d99aad92066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f9f1f3ffb9db5fb6d32d3f05b21a21d51257e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f9f1f3ffb9db5fb6d32d3f05b21a21d51257e0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
