)]}'
{"id":"openstack%2Fmistral~884577","triplet_id":"openstack%2Fmistral~master~Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e","project":"openstack/mistral","branch":"master","topic":"bp/add-maintenance-mode","attention_set":{},"removed_from_attention_set":{"35584":{"account":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"last_update":"2023-06-25 18:57:48.000000000","reason":"Change was submitted"},"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2023-06-01 13:43:18.000000000","reason":"\u003cGERRIT_ACCOUNT_35600\u003e replied on the change","reason_account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-06-25 18:03:11.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-06-25 18:57:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e","subject":"Added maintenance mode","status":"MERGED","created":"2023-05-29 09:05:24.000000000","updated":"2023-06-25 18:58:57.000000000","submitted":"2023-06-25 18:57:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884577-bp/add-maintenance-mode","meta_rev_id":"9799f6cae06806e64219346008dd3852b82f6b7b","_number":884577,"virtual_id_number":884577,"owner":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-25 18:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-06-01 14:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2023-06-25 18:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2023-06-25 18:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-29 11:16:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 13:42:18.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-06-01 13:43:18.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2023-06-01 13:43:18.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"e1f755d3eea869340c692b515b4e17546c5fd1fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-29 09:05:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"760d716b6949aadc146cefbffc8d081e22677d3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 11:16:09.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d12f816e44f84ceeacfe2f397f5ed461\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed38c18a1ae4488a756f2cb515f2054 : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5053e636cfd94e6494b411759c79166d : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27765b0b0bc9422eb89b3d3f2c28923e : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d022a12e0bb49a4999cfa7e2cef3f6d : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27199379e0654403ae13defef69cfb6d : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3ad84b0994d43d39a13ef6afed46b68 : SUCCESS in 15m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c952e7731a0940e2a5f5d1eee0bc596d : SUCCESS in 34m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58a8a12dde554cd184356a5528ec3b7f : SUCCESS in 41m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76ed7b40c1c641608afbbc2a7a94a81e : SUCCESS in 37m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe0a33e1199a4becbe1e40eba9e661a2 : SUCCESS in 41m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/120f7ee2150943fe802ab0d8085acab9 : SUCCESS in 33m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d23c79e3f1604fa3a917c9d8f3a06399 : FAILURE in 4m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f62b9c6534a848c6b06dfbcaaa21fa14 : FAILURE in 4m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5117bde4495543869d41f4115a9c1672 : TIMED_OUT in 2h 03m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4321f2514ae269bd16a4c19eb86a80f5d67a3a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-29 14:03:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4f0f427bd1214a13506dea2a2967a62a226763b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 16:09:30.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/516239f3b61e4acca619f754fe07694d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dac7f7dab4a4b4cacfccf6d375f1f30 : FAILURE in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c29e996a0454a6687ff6794ff46ca37 : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4ffd16ad8c4448d8fd25536f44459ac : SUCCESS in 11m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cb21a3706284308afc4c05fc23d2a12 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f78c4e98264410bb59a4c5e4ecf5bf : SUCCESS in 9m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de26e4962a3049cd894e883dec9f525d : SUCCESS in 14m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/590c349fb3374b709a462c24ffda3fd5 : SUCCESS in 45m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fde37882cbba49fda90936d66cfface2 : SUCCESS in 39m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f590bc4865b5455ea22f674236ba9a3e : SUCCESS in 35m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a7f9faa898544f0798a4cd1a734d2709 : SUCCESS in 35m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/40cf0a3709374626918282dc27975f92 : SUCCESS in 45m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8392d86bf76d4960ab968bc377aa5a80 : FAILURE in 4m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/82f5073438eb4beabf0ced842a992b0d : FAILURE in 4m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/aa5c6adf607141008ebbd9da9eb6dd80 : TIMED_OUT in 2h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b17d2e740c7081985b6ffedcdba9f893a0ff9cc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-30 06:35:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"201ec3d3facba64bafd0607a46ccf82efec375d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 08:43:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33881366863947d19c7e4c343776c49d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949a2be8d225482ca7d57c7a82e709db : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11429bff2fc147958afa827241d4d3cd : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1af9c8eab93b4b0ea17c26cc14ba7a89 : SUCCESS in 10m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1c142e3cf554e27b76f2c4b418d6f11 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631cebc9da1245399f3d1e2b847f2445 : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79476618879a4027a10f7178d3b78a1a : SUCCESS in 13m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5abee27cdae45ef8cbc8ea39d161407 : SUCCESS in 46m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbbff134d1064c00a94fe7070ca34ca5 : SUCCESS in 33m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc343de063584e95a07a4b6a4b0e3315 : SUCCESS in 36m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d03b7724a85c43efa54dab86ce5ba289 : SUCCESS in 31m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7be79c231fe547d8902239da3d241517 : SUCCESS in 29m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/20de7e570cfb454aa361c1a537c10acc : FAILURE in 4m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ec0cb36b67ab429e8052a6a721e589ac : FAILURE in 4m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1281b5b1c60349aa9f91a5aae1ae77ee : TIMED_OUT in 2h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c325eefb087f3a7e344ff191d6049f3473cf2d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-30 09:52:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"90c6ceec5940f38133b6eb2e3c6ba81ba33861da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:00:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d725956830e479f9538733e10c8eaae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16643e7b2c1541bc962906b79bac3e81 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d94733ea0af414fa4ed9aa036b9d530 : SUCCESS in 13m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65fdec0cda104585a7dd2340e57545e6 : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed925a02a05c417ebe89b990ffc17aab : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ef2f637b384e1fa0d45651d8ed1394 : SUCCESS in 7m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ab2291efc0b47a1856bf800c144ac88 : SUCCESS in 12m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4c669f1cb9dd4effbd643e89306848b6 : SUCCESS in 38m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90af788dc80041c09c61faea0e95a23c : SUCCESS in 51m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4041a174e4a94b94b5932757b3f458bd : SUCCESS in 37m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/956dc3135f0b4e6897669773947af1aa : SUCCESS in 42m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fe4179d361b74782a7544499d2e449b0 : SUCCESS in 46m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9b5a6b24149146e7a32f68901150c574 : FAILURE in 4m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2b0318bd8ecf4cdc9678aa2d0107894e : FAILURE in 39m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c8dc0de105d3473f80c8a346854a1d7e : TIMED_OUT in 2h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7ca97ac07ef3d01d17ef0399c4033694762b9774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 06:21:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"be9c1a2681abcde0cbc97245773f0fbf73135112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 08:29:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8182b465a71d450b99af72bcd2a7fe6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c41860ec7140a5abe93d0fe4ae8fd1 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d4d8e62bb3846718fbc409341170190 : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6ad193b17ae42969419c0c424c893ec : SUCCESS in 10m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aac8c70c135c462985cb647ada1050ce : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b65670b1b3e41ae9f330cbb250ef755 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a37ce77760ee44eeaf69aefc3b870be4 : SUCCESS in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bcf8fd2653e74282939f23b6d9369060 : SUCCESS in 34m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd7e5e4444042d4af6cfdb0e7e43c14 : SUCCESS in 46m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be6f02034de14a92a87d35ab60f8b2da : SUCCESS in 52m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6f16d6d1578f4605a860ee62fd006314 : SUCCESS in 34m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c1ac5c91ffaa45989452082b64c20e2e : SUCCESS in 35m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e4c3d21b85f14c22a33b65263427bc32 : FAILURE in 4m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/45f88ce6e5304c7996894dbe6066f8e4 : SUCCESS in 41m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/142bfbe30e5647e9aada1650d7f818ae : TIMED_OUT in 2h 02m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"995971abe564522cb72337912fdfabf73a70ffb4","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 09:15:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4473fe9329a40d17816d3e0e199f9e8451e887e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 10:58:38.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"84d2f330cb7ba3f08719548fbdb2d15060ec25a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 11:22:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"abca94dc4658032990b69b669e0c21796c7730fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 11:31:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"22aac1e2be400a925cd264cc57ab3fdface1e002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 11:48:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"abdb684f5508dc8437e5c201bd37861dc7b53791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 12:01:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7e66b737574816d9e30efe69aa553cce5ce29c0e","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-05-31 13:18:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"48ef2b14242127a4980ec1ecb1e9db8948f48dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 14:06:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31117d5962774f699b63a1eeec4b3187\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6f8ac24b8649ba977f33db3fd583cd : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b3d27e18740429ebdc908684fa70333 : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f46d32b0676d47adbffd0953f1309d69 : SUCCESS in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c84c14c4f28f4a929270b8533f3239f3 : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9490376c4c704f47a54d438a2927a37d : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43e210cc778b4d7db2194736f22e5605 : SUCCESS in 11m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e7764660ca7f4f4d8792814e5262459c : SUCCESS in 34m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9d9ed406fb94bd9b12fe96e23b26f33 : SUCCESS in 44m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22ebb3d7a08543a88aa434f5f8998dcf : SUCCESS in 54m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c166f7947dca414e91c7962834afb09a : SUCCESS in 31m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eb8780eff54949b8875c694d89535ae3 : SUCCESS in 36m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/11cfab40f873475cbee6c348b84afdfd : FAILURE in 40m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6e6d18f57797463c8c2367b0c0613a3b : SUCCESS in 38m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/44617bf09c3b4684b6e42d59ae7a1f14 : TIMED_OUT in 2h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2e646501344faa0d930db5b267480e8fa61f7f6b","author":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"date":"2023-06-01 05:58:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"20d21ae7ab5475221d1a89f51f9decb04cc14523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 08:06:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/017b6238955a495491b38061f31a055d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603f4e70fd4f43138b13fd86c929af36 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2243635e5554b779c173cff14918f96 : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d58cbf79f8c4216ad7460a8ede5fccb : SUCCESS in 10m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae6dcc6999bd42bc8a59033ede9b6e14 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cf4f97c39649328786ebd731f5e5c2 : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd1c4bb59264f7b9870b9a047e1f3d5 : SUCCESS in 12m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/112f15056be946d1bf9a55676ac03cc0 : SUCCESS in 31m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae7650e6e70c46a1a511b200e177c07b : SUCCESS in 32m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8461ccf7cee4d23a158bd866ab8678e : SUCCESS in 34m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/42565d3b47a040708f33b962d6452350 : SUCCESS in 37m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cc808788724c49e18ff0297190642893 : SUCCESS in 35m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ee930d6ef7b84fc7aad6f010c5c5ab47 : SUCCESS in 39m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a55a5f38db6c4839abc21fb569c047e5 : SUCCESS in 43m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/78860cf391fa4460a2b1b20589477b3c : TIMED_OUT in 2h 02m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"651356c003f8cca7e14f5558ee82d1f7a997c7d9","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-06-01 14:03:51.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7ef7e69ac934eae074dd0cae6c99486d409a3997","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-06-25 18:03:11.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"02016e2ef6a007cf0a866e27d046c57c394fe75b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-25 18:03:46.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"da21165f9a29aa0991246ed19a2daeafb57da563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-25 18:57:48.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b48788588a694dc38b6b110edc11f83b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badb62831a354b0c94f20cecd6f87b48 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e6c4d3e1ca145b8b94abdd61cbaf6ac : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee528f4657fb43aba0de971a7d3ccc1f : SUCCESS in 10m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c1c7acf08bd41ff90cdf7b47da2f9a9 : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5315dadc9ef41d9abf9ea462877363e : SUCCESS in 7m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6a390cd291a341b4a7db246ac451bc1f : SUCCESS in 48m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a9b1b31ff1d4718bf7717d3f7758514 : SUCCESS in 46m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef3c22c34a254446a5a4bb402b7d3d05 : SUCCESS in 45m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cf3fcf90e9f549e9b350e732014c013d : SUCCESS in 45m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/62b757585af64394afa63051f0a423e2 : SUCCESS in 49m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6002cb94d431425ba25111f4b7d62268 : SUCCESS in 43m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b3368f6ec7bd4bd7a2be096dad4b175e : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":10},{"id":"457f4b3828514ac391be43eeac20abbb8208317f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-25 18:57:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"9799f6cae06806e64219346008dd3852b82f6b7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-25 18:58:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2174860362144b78a196f8fc05d6d18b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46087f1302494b7a80b35d8c9b87793e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"51574ef6efb0167695727ef71d1c9eec7f2b81f4","revisions":{"5ed6ee8e9279674cc932d8d0874d2938bc204077":{"kind":"REWORK","_number":1,"created":"2023-05-29 09:05:24.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/1"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ed6ee8e9279674cc932d8d0874d2938bc204077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5ed6ee8e9279674cc932d8d0874d2938bc204077"}]},"branch":"refs/heads/master"},"0b7fbd905bbf5a0fad1a2b97f8662e399ee0b2fd":{"kind":"REWORK","_number":2,"created":"2023-05-29 14:03:58.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/2"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 14:03:43.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7fbd905bbf5a0fad1a2b97f8662e399ee0b2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7fbd905bbf5a0fad1a2b97f8662e399ee0b2fd"}]},"branch":"refs/heads/master"},"04e8d7292e1ad7cd08dd1ed83117b7af27f3e8e3":{"kind":"REWORK","_number":3,"created":"2023-05-30 06:35:28.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/3"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-30 06:32:54.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/04e8d7292e1ad7cd08dd1ed83117b7af27f3e8e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/04e8d7292e1ad7cd08dd1ed83117b7af27f3e8e3"}]},"branch":"refs/heads/master"},"15ed47c456120d0d8a903bbdf4265612ac2c5888":{"kind":"REWORK","_number":4,"created":"2023-05-30 09:52:59.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/4"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-30 09:50:56.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/15ed47c456120d0d8a903bbdf4265612ac2c5888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/15ed47c456120d0d8a903bbdf4265612ac2c5888"}]},"branch":"refs/heads/master"},"f8e0aa4ad573ffd72cd85f5b6ba514da00411098":{"kind":"REWORK","_number":5,"created":"2023-05-31 06:21:47.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/5"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-31 06:21:33.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f8e0aa4ad573ffd72cd85f5b6ba514da00411098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f8e0aa4ad573ffd72cd85f5b6ba514da00411098"}]},"branch":"refs/heads/master"},"7ec7e5082fbef979646111dafa7379e335844c0d":{"kind":"REWORK","_number":6,"created":"2023-05-31 10:58:38.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/6"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-31 10:58:21.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7ec7e5082fbef979646111dafa7379e335844c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7ec7e5082fbef979646111dafa7379e335844c0d"}]},"branch":"refs/heads/master"},"bb2872225a813cd7f8f6cf43462fbfd296addd05":{"kind":"REWORK","_number":7,"created":"2023-05-31 11:22:33.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/7"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-31 11:22:20.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bb2872225a813cd7f8f6cf43462fbfd296addd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bb2872225a813cd7f8f6cf43462fbfd296addd05"}]},"branch":"refs/heads/master"},"121dd741bcf5861991737e31f9b62660a8c9ec78":{"kind":"REWORK","_number":8,"created":"2023-05-31 11:31:35.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/8"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-31 11:31:22.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/121dd741bcf5861991737e31f9b62660a8c9ec78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/121dd741bcf5861991737e31f9b62660a8c9ec78"}]},"branch":"refs/heads/master"},"61bd2800da1bbd3803bee06f26567c5f36eb6e2b":{"kind":"REWORK","_number":9,"created":"2023-05-31 11:48:36.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/9"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-31 11:48:08.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61bd2800da1bbd3803bee06f26567c5f36eb6e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/61bd2800da1bbd3803bee06f26567c5f36eb6e2b"}]},"branch":"refs/heads/master"},"51574ef6efb0167695727ef71d1c9eec7f2b81f4":{"kind":"REWORK","_number":10,"created":"2023-05-31 12:01:53.000000000","uploader":{"_account_id":35584,"name":"Vasudeo Nimbekar","display_name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","username":"devnimbekar"},"ref":"refs/changes/77/884577/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/77/884577/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/77/884577/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/77/884577/10"}}},"commit":{"parents":[{"commit":"d6b7f7305d249a5beaacddc90bbc1003680fdb87","subject":"Merge \"Starting tasks via RPC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6b7f7305d249a5beaacddc90bbc1003680fdb87"}]}],"author":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-29 09:05:04.000000000","tz":330},"committer":{"name":"Vasudeo Nimbekar","email":"vasudeo.nimbekar.opendev@gmail.com","date":"2023-05-31 12:01:42.000000000","tz":330},"subject":"Added maintenance mode","message":"Added maintenance mode\n\nChange-Id: Ie42fcfa96044da1793f1afbff5fb2ae5534d5c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51574ef6efb0167695727ef71d1c9eec7f2b81f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51574ef6efb0167695727ef71d1c9eec7f2b81f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
