)]}'
{"id":"openstack%2Fmistral~886138","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I92f6d73ccfb67a633d1494f9b1c6990439924235","project":"openstack/mistral","branch":"stable/train","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-26 04:43:17.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I92f6d73ccfb67a633d1494f9b1c6990439924235","subject":"Update zuul queue configuration","status":"MERGED","created":"2023-06-15 11:13:48.000000000","updated":"2023-06-26 04:49:27.000000000","submitted":"2023-06-26 04:43:44.000000000","submitter":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"886138-fix-zuul-errors","cherry_pick_of_change":886137,"cherry_pick_of_patch_set":2,"meta_rev_id":"c32dd636c2cf06254d28d6ddf0cca600b2e04fd1","_number":886138,"virtual_id_number":886138,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-26 04:43:17.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-15 11:51:40.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-26 04:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-26 04:43:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 11:13:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-06-15 11:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b318861d6c6e12fdf71e78839f3c16907d66610","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 11:13:48.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"6d68954cc99e564671bb7982a4ff4917df7c8406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:14:11.000000000","message":"Patch Set 1: Verified-1\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":1},{"id":"8af378467f321127820c7e2a744a429cd9c7fb93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 11:14:27.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d14c568ac0741a2a79ade61d72fb33fb6dca3b6a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 11:14:40.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"172e7eb22b87ae7a4bca98f4a4eec9f4c9080fc7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 11:15:35.000000000","message":"Topic set to fix-zuul-errors","accounts_in_message":[],"_revision_number":2},{"id":"31f9b0e8aedb8ae5fb02abf638decb4423a50f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:51:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f804f65b12a482980b2b5f0c8e39941\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31003efaaee5493aa925dbeb70b4c896 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f4efa8439a5453e8799c2591d934f3b : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f11eb61a2204a809cb229060fed4748 : SUCCESS in 9m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fe4a6b5919e409587b409c109a2e7e8 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24bb037395e54cd89fe545b61898f0f7 : SUCCESS in 5m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dceaa25d371a476d8850116d4adce023 : RETRY_LIMIT in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b956670b2c14da0a2ffc5fdfc53e6c6 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0391db1784f446b1a9d31d29ebbd58d4 : SUCCESS in 8m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ccec38aa705447a2af12f8e872cb5626 : FAILURE in 31m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1dfe7fae95cb485c915e424b3e396cc0 : FAILURE in 19m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e0c9697ed2b34040b33a603f6d9c4f73 : FAILURE in 19m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0662532812d448b39fd0ee5351453c8b : FAILURE in 25m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0f244a8cf64c443f9c2ea35724e8bb57 : SUCCESS in 27m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6fd81ab13a5e43948952b23cb55235b6 : FAILURE in 26m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/12579e697e284a65a19e9a89f9a9a1dd : FAILURE in 23m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7af320724b4a48c6913a58fcba2edd12 : RETRY_LIMIT in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e8e35a84db659ce43aff5505b399e657eb2cfbf6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-16 09:31:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db3df3afb87d9714568bd92e7836ea0a2424f3ea","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-26 04:43:17.000000000","message":"Patch Set 2: Verified+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ba31f4b424fbe4557d8e236335e2837ad4934af","tag":"autogenerated:gerrit:merged","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-26 04:43:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c32dd636c2cf06254d28d6ddf0cca600b2e04fd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 04:49:27.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a7178204f044ce784af387302537b17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529732147bc640ff9502d2fc1cba1fdf : FAILURE in 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c66cf75a50bf2772b0b23f89f1a911d3d5f72875","revisions":{"42c1178fce9f8de9b83488d139cee30b51bedbff":{"kind":"REWORK","_number":1,"created":"2023-06-15 11:13:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/38/886138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/38/886138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/886138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/886138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/886138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/38/886138/1"}}},"commit":{"parents":[{"commit":"62b0b37e3a01ed12f6791d596a93babc698e7217","subject":"Merge \"Update requirements to include oauthlib\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/62b0b37e3a01ed12f6791d596a93babc698e7217"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-29 08:40:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-15 11:13:48.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I92f6d73ccfb67a633d1494f9b1c6990439924235\n(cherry picked from commit 13c015b3fd86e1373b8bbc9bb469330d67fd292c)\n(cherry picked from commit 78d225b32d221292c3458a605fb3e4da138e020a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/42c1178fce9f8de9b83488d139cee30b51bedbff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/42c1178fce9f8de9b83488d139cee30b51bedbff"}]},"branch":"refs/heads/stable/train"},"c66cf75a50bf2772b0b23f89f1a911d3d5f72875":{"kind":"REWORK","_number":2,"created":"2023-06-15 11:14:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/38/886138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/38/886138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/886138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/886138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/886138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/38/886138/2"}}},"commit":{"parents":[{"commit":"62b0b37e3a01ed12f6791d596a93babc698e7217","subject":"Merge \"Update requirements to include oauthlib\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/62b0b37e3a01ed12f6791d596a93babc698e7217"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-29 08:40:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-15 11:14:26.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I92f6d73ccfb67a633d1494f9b1c6990439924235\n(cherry picked from commit 13c015b3fd86e1373b8bbc9bb469330d67fd292c)\n(cherry picked from commit 78d225b32d221292c3458a605fb3e4da138e020a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c66cf75a50bf2772b0b23f89f1a911d3d5f72875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c66cf75a50bf2772b0b23f89f1a911d3d5f72875"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
