)]}'
{"id":"openstack%2Fmistral~886140","triplet_id":"openstack%2Fmistral~stable%2Frocky~I92f6d73ccfb67a633d1494f9b1c6990439924235","project":"openstack/mistral","branch":"stable/rocky","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-21 15:15:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I92f6d73ccfb67a633d1494f9b1c6990439924235","subject":"Update zuul queue configuration","status":"ABANDONED","created":"2023-06-15 11:15:01.000000000","updated":"2023-06-21 15:15:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":886139,"cherry_pick_of_patch_set":1,"meta_rev_id":"222d64ac2db608a3e66c0f28eb22cdbe205d9d88","_number":886140,"virtual_id_number":886140,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-15 11:57:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 11:15:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-06-15 11:57:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9577752f1f38a4f5c0ddaead25f18bc873a85df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 11:15:01.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"ed7f81414a4ffa56ec7da5c2f3fd69ff78ad9fdf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 11:15:35.000000000","message":"Topic set to fix-zuul-errors","accounts_in_message":[],"_revision_number":1},{"id":"89c408326e014ba370697111959b5bd28f5cea99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:57:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a68d6dfc039d4faebe20a9528f0f10b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a380b3bc90c402a862c915a25836f12 : FAILURE in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ab23e4297cc45409f38dc2e018720cb : FAILURE in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ade923639dc4486480857a1706a2723c : SUCCESS in 8m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/23ce6e17337749a2bc7329f2ab1faf8e : SUCCESS in 9m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/999a9ffedd0543f8aa5bf0b5715c30a0 : FAILURE in 2m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8dd59228b4d4888a6ea37720caa9d56 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bb65678896145a29dad8adc3ac22e71 : SUCCESS in 9m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d66654e7817d480ab3f4604f51122b5c : FAILURE in 11m 13s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e6399061680f49dd842cec7d75a010c2 : FAILURE in 25m 46s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2d96253e00b24be8b6e75db376ad984b : FAILURE in 11m 46s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fa68d672034d48ef9ffa3d32876709ed : SUCCESS in 29m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c0e33d12ffa7417fa31f918e33232fc3 : SUCCESS in 25m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c0cb518a5ce74e3882c53a193322c96e : FAILURE in 18m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aedf685590f34ad4bee5f829dc29a333 : RETRY_LIMIT in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"222d64ac2db608a3e66c0f28eb22cdbe205d9d88","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-21 15:15:09.000000000","message":"Abandoned\n\nstable/rocky of this project has transitioned to EOL. To be able to delete the branch the open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ea98ff4b9b142446203d5fd9816851edfd9784a","revisions":{"2ea98ff4b9b142446203d5fd9816851edfd9784a":{"kind":"REWORK","_number":1,"created":"2023-06-15 11:15:01.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/886140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/40/886140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/886140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/886140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/40/886140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/40/886140/1"}}},"commit":{"parents":[{"commit":"2968b8bbb743ada6df7e36c969228bef42830224","subject":"Disable the use of anchors when parsing yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2968b8bbb743ada6df7e36c969228bef42830224"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2022-09-29 08:40:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-15 11:15:01.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I92f6d73ccfb67a633d1494f9b1c6990439924235\n(cherry picked from commit 13c015b3fd86e1373b8bbc9bb469330d67fd292c)\n(cherry picked from commit 78d225b32d221292c3458a605fb3e4da138e020a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ea98ff4b9b142446203d5fd9816851edfd9784a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ea98ff4b9b142446203d5fd9816851edfd9784a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
