)]}'
{"id":"openstack%2Fmistral~893847","triplet_id":"openstack%2Fmistral~master~Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-15 18:28:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5","subject":"Allow admin to retrieve all values","status":"MERGED","created":"2023-09-06 09:22:11.000000000","updated":"2025-02-15 18:29:24.000000000","submitted":"2025-02-15 18:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893847","meta_rev_id":"d2c9f48607ce7c4424147b59f28d0faff4781ecf","_number":893847,"virtual_id_number":893847,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-15 18:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-15 18:03:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-15 17:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-15 17:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}],"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 10:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 15:41:41.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"},{"updated":"2024-11-15 14:52:48.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"CC"},{"updated":"2025-02-15 16:03:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-02-15 18:03:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"c8774ce66e21f82407709b059ea77b6e4e81f481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 09:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e15cc8374585eb5b201b39652aa3b68bb4ee86f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:06:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d301de0d766c43fd942f2e0f9443d46f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef226a51bb4840deb1ff2fcfbb9a8bf0 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22ca9de42d7a4bf6951dac9565fe0db4 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e47547e265f9410caf9512369b90da13 : SUCCESS in 13m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ec5c91b782848e3a7c999ac73525287 : SUCCESS in 9m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cb52307896f404f825ff2cf2f25d4c1 : RETRY_LIMIT in 1m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b283ed123d4b68944196dff7026eed : FAILURE in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/992acac2722c44568705846488f828ae : SUCCESS in 16m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/41b0ba5216dd45658d0e5e5ede23b02a : FAILURE in 28m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c3569862ce048caa755520eca1cc617 : FAILURE in 28m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa3fae033e4d475798787a8fcfdd87d4 : FAILURE in 24m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9c43197f19cc46409c10ecbd6271e7dd : FAILURE in 38m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ac1fbd80d5bd423eaeedb572148dd87c : FAILURE in 27m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8391cf85768240349cab4e42a6cf44b2 : SUCCESS in 38m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9fb6df27d48b4947a397220198f4cc8d : SUCCESS in 41m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/832df19226cc49d2a8e1a86c29678312 : FAILURE in 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d06c37e516e0713a2fd1e7b5d2a5e15e62397b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-25 18:32:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8f2c1879c5f7be9813b2105ee767b277d89edaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:24:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f45548c55f6641429df15e94b7e812c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f25bfd6e1c462bb3c73bfb431a4574 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d741e9b31a1145a492882adc3eb77d50 : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1177babb05a44791a50faad9ec6e3c58 : SUCCESS in 8m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5784f954f8843e7aef90e158b2dc688 : SUCCESS in 10m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa25e75bfadc43b7a1a84be818497897 : RETRY_LIMIT in 2m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ea3c70cceb43539161c65117978d84 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba401a221cec45edaea5e9cae26edb45 : SUCCESS in 15m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6bd6378793f94989bf6edf745e9ec871 : FAILURE in 36m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1125372a08b2437da86aeb9e0bd391a3 : FAILURE in 19m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb1d00b0987c4ada9db749d918a372cb : FAILURE in 20m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/76cec33aad73432281a86893e1eaffee : FAILURE in 19m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/38114db216a34bafb091021012944fee : FAILURE in 19m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/21712120709e4c50b95a2fa078b75e37 : SUCCESS in 44m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3409ef8834cc4c7388eaf992323d4b83 : SUCCESS in 35m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f32dee923982436297ac429f68528673 : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bacbb237cfac49d8440c4de34fc6c48bdc01bbe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-09 21:57:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e9d88123458caac22e1bf75ed3f3163672c546e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 00:06:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5f6c88c560412c8f98da78be35ca80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a3bf8c93df4129abcb793bb8e5131e : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e46a9ffce34c33a10a9502130407a5 : SUCCESS in 16m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d39f5c60e71a4eca879379a0b449221a : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b018099bae6b4bf0b53e80324bdf22ee : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08cbe1576be84f64a6de621bf010d574 : SUCCESS in 15m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/005278c8fbe54a0180aa89ab95b973ea : SUCCESS in 28m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11ea3542cc3841e5a32cf7c44de88add : SUCCESS in 38m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44f4debdd546479fa1a220b61c023ed9 : SUCCESS in 37m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1a29db281d594529ba4316d38ddfaff3 : SUCCESS in 38m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6ae14d6cd3b64a739ae8099dfd54deed : SUCCESS in 35m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5335794b02344497a01c21d9b9df997a : SUCCESS in 45m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/304dd0876f7849749e55b5e1b2659062 : SUCCESS in 38m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/19cd55fd661f477fa549ec285d92e398 : TIMED_OUT in 2h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a70d4aa5809ba3495824623ea2cf3b67e7ed636","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 15:16:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4fcacf65236bae0822ca43a477606a98ff5337a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 15:16:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"121764c1bcb506789630ef990cd05015d09919e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:23:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/103f87edb0454415b50a7732b734dc29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6adcb0088d48ff97385ad7a0172bcb : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13a771c1b53b4e0bb14233cc4faf207c : SUCCESS in 14m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c71265602a44838af55d5710cee2a4b : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec4ae378ab74d12acad8e40e1bcf88a : SUCCESS in 7m 16s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/86dba2e3a5be4a7e9f30b42af8f0a408 : SUCCESS in 37m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4283cea551640a5ac793791b98a382f : SUCCESS in 43m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e458f813c37446378ff83966d9529433 : SUCCESS in 39m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/67974037b5524ac6b00802bad7adf10b : SUCCESS in 46m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9d493e0266924517aa49795fa2fbed40 : SUCCESS in 57m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/61c9eea29a974b3cb12628e2f148cc82 : FAILURE in 51m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a836045f6139482e93ebd1fc5728856e : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":3},{"id":"fbf1e0659b14993c3dfa6c2daf7a5c5f93aa0280","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-02-22 15:41:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c03c9877f009f0a5aa0661c99e95405463cd8ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:50:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4532d31795e43bd8b696e5f13ae245a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8747c6a95e48ddacbbf36c248905a7 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2c57491de344561a96c5197ed1dc8d1 : SUCCESS in 14m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8823cfaf6bc44739e8b8858446051a9 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f571fb19a3d8435c8e2e4b999689761d : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cd66fb1d85b41b0bad903fd234aa5de : SUCCESS in 18m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1d1d5c9019fd49b2840b4f953f213865 : FAILURE in 49m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7219a05085ee4adc91de664f582a5ab9 : FAILURE in 40m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f022d8e34cb5432bbd0b53fac20056d1 : FAILURE in 38m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/24ee3554682b4812848e0cc63957ed5a : FAILURE in 35m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2cda6bc165ee46369ba4c9eebe2e038c : FAILURE in 46m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ef1efb223d9e4fe2b5d7ef6146b6333a : SUCCESS in 58m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/abfdc501a24a4251908b4bfe4e6f41e3 : SUCCESS in 53m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5b07d0b5e4224763a3a8679b99f7117b : TIMED_OUT in 2h 02m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"53070821a98deb1692e8094a8a65e729abd6ef1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-07 21:25:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3cbd2ce6e9d86a527bd1d603d1e0f0182aa5969d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 22:11:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0483c1befb3447039a3b9a95daa4e957\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40be23e2acd546bbb00e7ca67f98cd8b : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3a6af3f6c944d25acf508c71bc70aef : FAILURE in 14m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f942fda263a4524a3979604bfefd6f2 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a1db40d4c349f4b61412e96b4a3a23 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767ed7a886dc4b8b9f55570a2f63890e : SUCCESS in 12m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5ededf1086bb4debb220eac1974d3c5f : SUCCESS in 39m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/157626108d51473893ed760fab0768ed : SUCCESS in 27m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aefc4de1ed8f4c449da235b9ab8e8e98 : SUCCESS in 38m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a7c08f54a3ce42219205820e45ba4bef : SUCCESS in 29m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/feb2b1d7c5d2426f8a58e1c2872e4491 : SUCCESS in 28m 25s","accounts_in_message":[],"_revision_number":4},{"id":"8f4b289f42a52a288e72dc2c66444469b0c4066e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-31 14:46:40.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b09095dbd6a7e04b1b72381a544c8cb914ca7a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 16:02:58.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_11583\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"_revision_number":5},{"id":"15bd3fa2afeb7f08a4978fc0d61e2ff8d4ca377c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:53:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a92e1de1eabe4f9da2169524d592bf0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc1e221dfa144f49cd21b9d978ea9b7 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9491699da5354c9092d405c96172d15f : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c278ba33a4fc41938b3dfb990ffb3387 : SUCCESS in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a23fe035cf46a39fc3212124896584 : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85093a85605644bfbd3f6ab9e59feea9 : FAILURE in 16m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a18ee04bdf384eafb18bd59503519e49 : SUCCESS in 44m 47s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/07c1970ee42640159081c278af34e636 : SUCCESS in 22m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f5dcdb5494e4fff9f545894b520a945 : SUCCESS in 29m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c63c5fa4ea874002887231d538f1ed2e : SUCCESS in 28m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c3cceece59d54291ae9c2f0225fbd7c7 : SUCCESS in 25m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3cc37845f0184417a550259dcab7aefe : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":5},{"id":"c862c1e20f107bd06842d6ac4ba2c42aacaaf2df","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 17:57:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7732cfdb6f6a2e3ef101c2601ebe1ea53c7b4319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 17:58:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fad1c565a84ad8a96d0766aec2917adc670771ba","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 18:03:23.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a479595c084ffc401f8d0735a3e774e9b8b0419f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 18:28:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c7bfcb21414358b6c9d1da1a523bd1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7017e5c84909481fb12f1485c7c36f06 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b83a5b03cce41c0983e309da49036cd : SUCCESS in 8m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d23cf05f31c4bf18e83f8c8ca6ffd2b : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e926e6f5534b769a4cdfee6773d3a9 : SUCCESS in 5m 08s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/252f4ede02a540cd917c7b8d7451a023 : SUCCESS in 23m 42s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/9c46acc3eb0848e09f57311fbcde93a4 : SUCCESS in 21m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f855706ceb754baeb552da4bd14e57bc : SUCCESS in 24m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b82cf18ecc944c9e969a800bf4c519ca : SUCCESS in 22m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9ebdeae4f6c14da2b9beb4136581a7d0 : SUCCESS in 27m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8a024204fd1d4e67afe24bc9f673cbdb : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":5},{"id":"49a14c7e900caeb82f4cd4ab16c27926f822962e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 18:28:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d2c9f48607ce7c4424147b59f28d0faff4781ecf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 18:29:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c9a5f825094bf09a28cbecbc224a61\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c277cefe4641158a7251555cb96999 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9bba73e0a56c84081c26909e652732e1e91081c2","revisions":{"35205140cdd57849d55c26e90fe3ddf63beb59ec":{"kind":"REWORK","_number":1,"created":"2023-09-06 09:22:11.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/47/893847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/893847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/893847/1"}}},"commit":{"parents":[{"commit":"f35d366794c9d42bba5207c57d8fafb5cb279e22","subject":"Merge \"Update get execution\u0027s tasks functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f35d366794c9d42bba5207c57d8fafb5cb279e22"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-05 15:41:38.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-06 09:21:21.000000000","tz":120},"subject":"Allow admin to retrieve all values","message":"Allow admin to retrieve all values\n\nSome of the admin calls are done without providing the project_id,\nanyway when we are admin, we want all, it\u0027s not needed to check if the\nadmin provided a project_id.\n\nThis allow commands like:\nmistral task-list \u003cexec_id\u003e\n\nto actually show the tasks.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/35205140cdd57849d55c26e90fe3ddf63beb59ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/35205140cdd57849d55c26e90fe3ddf63beb59ec"}]},"branch":"refs/heads/master"},"2dc83f4933346f7b84e663cd5494a1979d801525":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-25 18:32:19.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/47/893847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/893847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/893847/2"}}},"commit":{"parents":[{"commit":"7baff017d0cf01d19c44055ad201ca59131b9f94","subject":"Merge \"Fix test not waiting for action to be created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7baff017d0cf01d19c44055ad201ca59131b9f94"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-05 15:41:38.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2023-09-25 18:32:19.000000000","tz":0},"subject":"Allow admin to retrieve all values","message":"Allow admin to retrieve all values\n\nSome of the admin calls are done without providing the project_id,\nanyway when we are admin, we want all, it\u0027s not needed to check if the\nadmin provided a project_id.\n\nThis allow commands like:\nmistral task-list \u003cexec_id\u003e\n\nto actually show the tasks.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2dc83f4933346f7b84e663cd5494a1979d801525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2dc83f4933346f7b84e663cd5494a1979d801525"}]},"branch":"refs/heads/master","description":"Rebase"},"53da1ba8f29103caa53610312c6ef0c906e724d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-09 21:57:02.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/47/893847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/893847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/893847/3"}}},"commit":{"parents":[{"commit":"5be3de2fd4c1abccbecd768fd2d74061e9d14c4d","subject":"Revert \"Temporarily disable failing jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5be3de2fd4c1abccbecd768fd2d74061e9d14c4d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-05 15:41:38.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2023-12-09 21:57:02.000000000","tz":0},"subject":"Allow admin to retrieve all values","message":"Allow admin to retrieve all values\n\nSome of the admin calls are done without providing the project_id,\nanyway when we are admin, we want all, it\u0027s not needed to check if the\nadmin provided a project_id.\n\nThis allow commands like:\nmistral task-list \u003cexec_id\u003e\n\nto actually show the tasks.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/53da1ba8f29103caa53610312c6ef0c906e724d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/53da1ba8f29103caa53610312c6ef0c906e724d5"}]},"branch":"refs/heads/master","description":"Rebase"},"d62c418955bee931ea4e7985cb6ebea508574613":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-07 21:25:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/47/893847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/893847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/893847/4"}}},"commit":{"parents":[{"commit":"ea86527633bc421d55f9ea90d8f62b934d507716","subject":"Make mistral reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea86527633bc421d55f9ea90d8f62b934d507716"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-05 15:41:38.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2024-10-07 21:25:10.000000000","tz":0},"subject":"Allow admin to retrieve all values","message":"Allow admin to retrieve all values\n\nSome of the admin calls are done without providing the project_id,\nanyway when we are admin, we want all, it\u0027s not needed to check if the\nadmin provided a project_id.\n\nThis allow commands like:\nmistral task-list \u003cexec_id\u003e\n\nto actually show the tasks.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d62c418955bee931ea4e7985cb6ebea508574613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d62c418955bee931ea4e7985cb6ebea508574613"}]},"branch":"refs/heads/master","description":"Rebase"},"9bba73e0a56c84081c26909e652732e1e91081c2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-02-15 16:02:58.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/47/893847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/893847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/893847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/893847/5"}}},"commit":{"parents":[{"commit":"c14c59248b3f873f0b75a868d158f57db792bc5d","subject":"Stop using eventlet to start services in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c14c59248b3f873f0b75a868d158f57db792bc5d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-05 15:41:38.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2025-02-15 16:02:58.000000000","tz":0},"subject":"Allow admin to retrieve all values","message":"Allow admin to retrieve all values\n\nSome of the admin calls are done without providing the project_id,\nanyway when we are admin, we want all, it\u0027s not needed to check if the\nadmin provided a project_id.\n\nThis allow commands like:\nmistral task-list \u003cexec_id\u003e\n\nto actually show the tasks.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ida3e0c63005a3719f982dc1b9a4b47775f7d20c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9bba73e0a56c84081c26909e652732e1e91081c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9bba73e0a56c84081c26909e652732e1e91081c2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
