)]}'
{"id":"openstack%2Fmistral~894968","triplet_id":"openstack%2Fmistral~master~Ibb5297d4a25a53511d8e250b17a3db9dcecd2596","project":"openstack/mistral","branch":"master","topic":"release-health-check-cwr-bobcat","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-12-14 12:04:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibb5297d4a25a53511d8e250b17a3db9dcecd2596","subject":"DNM: gate health test","status":"ABANDONED","created":"2023-09-13 19:19:08.000000000","updated":"2023-12-14 12:04:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed5de30e0075640e6e4c04bbd927c668a131bb49","_number":894968,"virtual_id_number":894968,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-14 01:11:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 20:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f58455271d831e7529eb1133e8efb307b7e0544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-13 19:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f258155b62f628df1e4fc199fa4c50eaa9b6062e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:08:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eeea7718a35b4b04a00cb5978aa91844\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ebf93f491740e9b35e97878a49d9bd : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd43728f5804d91a9b54bd10b2673b4 : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1fc83fbd9d940b38266d3f7bccd407d : SUCCESS in 10m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e309ceebe0c4393b396a39d2cf94229 : SUCCESS in 9m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/370c37bd4eae4bc0899d727c49874a40 : RETRY_LIMIT in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad5ee1c8f6642cabea628b1d6db3ad7 : FAILURE in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9976612890484154b6d9b9d6327f39e0 : SUCCESS in 11m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3f4530e0665048af91ff9caf0cb71b12 : FAILURE in 24m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58891282bdd94dfb95fe047840cf1c04 : FAILURE in 23m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82a4c5302e874ebea42122612aff42a3 : FAILURE in 25m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7ff7d2884e9c42f78da600b8c5f0b018 : FAILURE in 27m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bdc0f42951614432a1c2eebc78a95986 : FAILURE in 27m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9136c221e404264a89208b79cadcbb3 : SUCCESS in 44m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d80af75a1c0f4e77a2d2f053b9d71113 : SUCCESS in 41m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a2be57dc293a464db16807d1da946cb7 : FAILURE in 38m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1c911bdf16890df41ddf6dfc6527b39f3d3a0e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-13 23:07:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_17685\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"_revision_number":2},{"id":"c53a045100894f657fde7857aefcf756bb34a780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 01:11:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58e12c6bb0b04b2bb77e8ee974ef432a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5368f0f31742458f3ff79bfcc61d67 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9df405703bfb406daba4ed36fa64c7ae : SUCCESS in 14m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/021e026d5b5941c8b6b82762a50ca585 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/145b3a2dde444522adca760cfa6b464c : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6950b4ef7374a10ad45cf60d42de13d : SUCCESS in 16m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9e95e3f8946c4114908fe71d17774822 : SUCCESS in 36m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3cc857ec59f48b39624afc68d1438ba : SUCCESS in 37m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4a9f311ca53492789dbd8bfea4d883f : SUCCESS in 36m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2d240dc3353e4253a78b36ecc4e44f02 : SUCCESS in 31m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e31614e0c6af49f2a8da97a260eed136 : SUCCESS in 33m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1b04cc27fd834ba2b5e0e82b4dc54e53 : SUCCESS in 42m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/83f6bb902de84429ad62dceb92296831 : SUCCESS in 39m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4d0c7462effd4906b8d0febcb655c164 : TIMED_OUT in 2h 02m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed5de30e0075640e6e4c04bbd927c668a131bb49","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-14 12:04:58.000000000","message":"Abandoned\n\nIt looks like gate is passing. Let\u0027s abandon this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68c18d286fee9205fd0e1173127101d5a3f17058","revisions":{"40202e5024b20ac23899ccddf04ae5b05ab7568d":{"kind":"REWORK","_number":1,"created":"2023-09-13 19:19:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/894968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/894968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/894968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/894968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/894968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/894968/1"}}},"commit":{"parents":[{"commit":"7baff017d0cf01d19c44055ad201ca59131b9f94","subject":"Merge \"Fix test not waiting for action to be created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7baff017d0cf01d19c44055ad201ca59131b9f94"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:18:58.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:19:06.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Ibb5297d4a25a53511d8e250b17a3db9dcecd2596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/40202e5024b20ac23899ccddf04ae5b05ab7568d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/40202e5024b20ac23899ccddf04ae5b05ab7568d"}]},"branch":"refs/heads/master"},"68c18d286fee9205fd0e1173127101d5a3f17058":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-13 23:07:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/68/894968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/894968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/894968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/894968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/894968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/894968/2"}}},"commit":{"parents":[{"commit":"cfbea4f743c04d8271584614fcdccf121e44924b","subject":"Merge \"Avoid RuntimeError caused by iteration over sys.modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cfbea4f743c04d8271584614fcdccf121e44924b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:18:58.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-12-13 23:07:16.000000000","tz":0},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Ibb5297d4a25a53511d8e250b17a3db9dcecd2596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68c18d286fee9205fd0e1173127101d5a3f17058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/68c18d286fee9205fd0e1173127101d5a3f17058"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
