)]}'
{"id":"openstack%2Fmistral~898981","triplet_id":"openstack%2Fmistral~master~I872f73578eb7622cd3d18db29544b5420f55121d","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2023-11-24 13:34:01.000000000","reason":"removed on reply"},"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-11-23 07:05:43.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-27 15:52:03.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-27 17:59:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I872f73578eb7622cd3d18db29544b5420f55121d","subject":"Temporarily disable failing jobs and fix bookworm bindep","status":"MERGED","created":"2023-10-20 19:18:53.000000000","updated":"2023-11-27 18:04:01.000000000","submitted":"2023-11-27 17:59:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898981","meta_rev_id":"14b1a4c50cd0fcf90bcbc50b39bebb1de337a987","_number":898981,"virtual_id_number":898981,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 17:59:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-11-24 13:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2023-11-23 07:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-11-24 13:34:22.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2023-11-23 07:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-11-24 13:34:22.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 20:19:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 08:46:43.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-11-13 16:53:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-14 19:55:33.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"e11d7844797a3deac3383b818f6475ccbf97c3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-20 19:18:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d1a334a5114f96453faeecdc3b5fb7d98dec7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 20:19:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3009e1d578374789a1e54f9a533019e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb2dc7b1eac48cd8f71f15e6998bbf7 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdbba3a5bde648ffa8defdc3d36811aa : SUCCESS in 13m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c439b247e99845ea9bfac905f96e84db : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c449190c137f4dceae850cffe67d1249 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b68d5d24add6424b873f6075a75205a6 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/988ffe57f06f400d926d2e594443f385 : SUCCESS in 12m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/96c497f0b5ea4163a4f1ff100c13ca81 : FAILURE in 25m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e02f30b745124e16870040e8bbb3b9d0 : FAILURE in 23m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c47b49bade34ca1ad2ce276b64b6f8c : FAILURE in 35m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/31b4acfe70bb4dc4b6c2f09e87f8a7ac : FAILURE in 47m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/80c1614c9d064bfb887d72a15d428e41 : FAILURE in 35m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2d824e25f4f440c0b8d698142901758f : SUCCESS in 55m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1cddd125f3b54e318ed9cd43de95dc6f : SUCCESS in 52m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ccc9d86395814e82a0093532d4c2a69b : FAILURE in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cc663ce8b5605847b039873ea92b1f22692746d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-09 22:32:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"02ec436e045f7e6edac2487d2a6f148cf378862b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:35:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d30a50cf197840ba95145e4ab4d80618\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24812a2b2f6143b49e88c40dae66776f : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0978a674eeb64fa68ee5c19fd629ea35 : SUCCESS in 11m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25fc2d43b432483fb9750503fd6346f1 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b4ac7209c5477aad440ab01ce542bc : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3954d1879ce473bb9fa3c9ee5e1ec50 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/912becac79de453b8005e4fde88f899e : SUCCESS in 12m 54s (non-voting)\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2dedb1ff2044ecf97fcb4694c439c62 : FAILURE in 39m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a72ae2695db94f70b04c6cf6df4f91a6 : FAILURE in 23m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8aa84d5544934e29a9cb5fc58196aba8 : SUCCESS in 54m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/35d15b7c662a410b8ba6b8d120d39070 : SUCCESS in 42m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0026fc428dbc454f8ed87a2fd28bc96e : FAILURE in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85954c1c238b210ce92e535255c65bb7df33252e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:05:21.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"58e98cf208c8176ae7a99eea2d3158de6d82dbb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 21:08:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbfec858e817474387f82b3549ef11c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b3ead8a7be4f669f90eccd59690f08 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5406b885ee5d421294140db8f5782488 : SUCCESS in 13m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60c0c212f9384ec294cced0cc7774c25 : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb1d58014fd4b1d805eb175848d3d5e : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c5c87cbecdf4c1294fcfb86baac5270 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bc81e49422d4090a4d522118698a60c : SUCCESS in 16m 03s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dd08575fc779478294ddf3483da8b515 : SUCCESS in 54m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a7c610042c3348c18bb5d6d69e84b6a3 : SUCCESS in 41m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/00dd9834330042f88e2b636dbb629ee5 : FAILURE in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e7401c856544c91d83acd27e3140b5babe32bc52","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-13 08:46:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ae5b56056ebd69aa5bbc9c72bbdc06d234fd99aa","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-13 08:47:51.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bebc2661ce190b68b17092675b9841e24390fea4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 16:53:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3b511d3d640d53acb7d2e33494990773fa86155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:52:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2762c76a548c890abaa6c2c6b71225086405d5e1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:53:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"991fc2d94dbdb1f4ac099587575080a7dc9affd0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:55:33.000000000","message":"Patch Set 4: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e3ef1892a4efba4e73f64c7dce23e057ad6e2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 20:56:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2df87926109d4a27aff3827ae42c1ef1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c2ccda3946406ba47bbdc099553256 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a179c1caac524425b2d46668bc3acad9 : SUCCESS in 12m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3672a19c73904403804452bf35502974 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e745a4981cf48eeabd555c30de5ceb7 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9800eaf3c32045128c1ba8a95b2427d9 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1490d38a8cb646248370427a09d8ee48 : SUCCESS in 19m 32s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0c7c50922e824fb5941dd451c8bf5d86 : SUCCESS in 56m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7e71ff0f0c4b47c9ac5b6ac3f764097a : SUCCESS in 39m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6c6ac33f769b4a6c855a13740302a4d9 : FAILURE in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"300fb0e579dc695aa6022e5c18bac8e27e31b591","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-21 14:25:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e757e6c7ae3ded43f2c2c30bb9436a2863880f3f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-23 07:05:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24aff16e15e54f082f4b2100342e48c7746cf555","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-24 13:34:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6aeb086839d9a6a86e0b9a7243f7e790852440d8","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-24 13:34:22.000000000","message":"Patch Set 4: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"c5e06a118b62556147bb68253358b8ad781ad173","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 15:52:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4b675320ebd42d9345e38c1a16b1cb3c9c78b4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:55:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23fc7fce8f2b4b3fb09a0ec8e2d9ae63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/142abf4cbec344d7b94b3094e760ce27 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de4e931b6c84392b2807160bb14bcf6 : SUCCESS in 11m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d49e8607ac4540e194f5d391f9d33cee : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7743b3a1194048a73afa8147c5fbcf : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b84cad0289e4c66aee169866e14f973 : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f49c1a5706944d0ba06caf20810b42d : SUCCESS in 15m 30s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/176cd8e801544b86b4bc352dc042ae8c : SUCCESS in 54m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/83a888f6aa254461873651618f2146e7 : SUCCESS in 39m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4056312da26345f68c95671302ca3cc7 : FAILURE in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0b727d214befc685dc516658b83e6384948ed563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:56:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c0c8395e4e5eba398ea447ebf3ed4ab48611e9a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:59:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/965183d153b74d37970e8504a6f82f49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d604354e17d748d68f8c19e484fe2d74 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08a80234a934aa2b53f1a01f2028ce9 : SUCCESS in 14m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f6aa4321c2c410a896b675ce3e9c203 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e75aace5cb7484aa025de694c1ec3cc : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/052b2cde49764142b6a4668db407e832 : SUCCESS in 4m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/32c4c49758754125827814824cbbd2fc : SUCCESS in 55m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bfd1f7304a394e76a575ffdebd5e4f87 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":4},{"id":"db27e8ae22138991f1a3977f7fd06a2bb796e2f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:59:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"14b1a4c50cd0fcf90bcbc50b39bebb1de337a987","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 18:04:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5250442b3a8449f697e397616beac5b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8e439d4132404a86d897de350b4ad9 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e0b1e63b768407d9d6ff1dbf2866f0f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ae0289e029757287fff787ba302fc8cb465bbc95","revisions":{"6480a2c488c568aa2ef13dd7dd89a590170ea1b7":{"kind":"REWORK","_number":1,"created":"2023-10-20 19:18:53.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/898981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/898981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/898981/1"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-10-20 19:18:24.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-10-20 19:18:24.000000000","tz":120},"subject":"Fix bindep for Debian bookworm","message":"Fix bindep for Debian bookworm\n\nDebian prefers Mariadb instead of mysql.\nReplace mysql-server with mariadb-server.\n\nChange-Id: I872f73578eb7622cd3d18db29544b5420f55121d\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480a2c488c568aa2ef13dd7dd89a590170ea1b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480a2c488c568aa2ef13dd7dd89a590170ea1b7"}]},"branch":"refs/heads/master"},"81454ec251f053fdf4ccc8615ba374ee4e5bf37d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-09 22:32:54.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/898981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/898981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/898981/2"}}},"commit":{"parents":[{"commit":"113ff41dcdfd31c4d63618078ddfcc7d0391a6d6","subject":"Temporarily disable failing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/113ff41dcdfd31c4d63618078ddfcc7d0391a6d6"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-10-20 19:18:24.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:27:18.000000000","tz":60},"subject":"Fix bindep for Debian bookworm","message":"Fix bindep for Debian bookworm\n\nDebian prefers Mariadb instead of mysql.\nReplace mysql-server with mariadb-server.\n\nChange-Id: I872f73578eb7622cd3d18db29544b5420f55121d\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/81454ec251f053fdf4ccc8615ba374ee4e5bf37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/81454ec251f053fdf4ccc8615ba374ee4e5bf37d"}]},"branch":"refs/heads/master"},"e324c4552daa9b94225e1223e4c008380cc9d470":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-10 20:05:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/898981/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/898981/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/898981/3"}}},"commit":{"parents":[{"commit":"06f9808da37b3ca9cf551cbd18f03e93127faa18","subject":"Temporarily disable failing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/06f9808da37b3ca9cf551cbd18f03e93127faa18"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-10-20 19:18:24.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:02.000000000","tz":60},"subject":"Fix bindep for Debian bookworm","message":"Fix bindep for Debian bookworm\n\nDebian prefers Mariadb instead of mysql.\nReplace mysql-server with mariadb-server.\n\nChange-Id: I872f73578eb7622cd3d18db29544b5420f55121d\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e324c4552daa9b94225e1223e4c008380cc9d470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e324c4552daa9b94225e1223e4c008380cc9d470"}]},"branch":"refs/heads/master"},"ae0289e029757287fff787ba302fc8cb465bbc95":{"kind":"REWORK","_number":4,"created":"2023-11-14 19:52:00.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/81/898981/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/81/898981/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/81/898981/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/81/898981/4"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:25:12.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-14 19:50:56.000000000","tz":60},"subject":"Temporarily disable failing jobs and fix bookworm bindep","message":"Temporarily disable failing jobs and fix bookworm bindep\n\nUntil we fix the mistral-devstack jobs that rely on wsgi, temporarily\ndisable the failing jobs.\n\nAlso, debian prefers Mariadb instead of mysql.\nReplace mysql-server with mariadb-server.\n\nChange-Id: I872f73578eb7622cd3d18db29544b5420f55121d\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae0289e029757287fff787ba302fc8cb465bbc95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae0289e029757287fff787ba302fc8cb465bbc95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
