)]}'
{"id":"openstack%2Fmistral~899169","triplet_id":"openstack%2Fmistral~master~Ieade7a30a192910fc3870b173017efd481f0b18f","project":"openstack/mistral","branch":"master","topic":"job_stabilization","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-11-23 15:53:46.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},"25733":{"account":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"last_update":"2023-11-27 17:59:07.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-27 17:59:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieade7a30a192910fc3870b173017efd481f0b18f","subject":"Add api_version for etcd3 coordination URL","status":"MERGED","created":"2023-10-24 13:25:08.000000000","updated":"2023-11-27 17:59:07.000000000","submitted":"2023-11-27 17:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899169-job_stabilization","meta_rev_id":"7cbed6b6b069cd18099748e4627149449eed6695","_number":899169,"virtual_id_number":899169,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 17:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-11-14 19:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2023-11-23 15:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2023-11-14 19:52:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2023-11-23 15:53:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-24 14:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-24 14:47:39.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"CC"},{"updated":"2023-10-25 15:58:53.000000000","updated_by":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"reviewer":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"state":"CC"},{"updated":"2023-11-10 10:30:17.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-11-11 19:33:50.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2023-11-13 05:36:22.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"88985f5bb36a9e836291a6af1513cf3e715d17e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-24 13:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e8c89568cc4df14e6fb0e373168cccdbf168aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 14:23:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/856c4ce40c714e7cbdae8fbabd0dd9f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76ffac06fb14c9cb99dd802d4727da1 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60376881bb545f3a72b3a7dadfb30a0 : SUCCESS in 10m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c6a4cecde274c158e0f1347b5d894b6 : RETRY_LIMIT in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07bb594b1821444996b9a5f93cc8695c : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/303c637de42d4b94a29a6306d393d052 : SUCCESS in 13m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cb0dab5be0a041db809363d2501a566a : FAILURE in 29m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bebfd1696c494837838a18f7b8833d97 : FAILURE in 33m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d20422705f4d4838956b6183b01b62bc : SUCCESS in 48m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/91b084e4aeec4d2e9a70939e18426ff6 : SUCCESS in 50m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/70611ebaf34e416b810bfe33c2009ad8 : FAILURE in 47m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d7ccaa8f2f284bdaa4d4095c83fe36d1 : FAILURE in 55m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d5fcc1adbea7473f868224adff17663c : SUCCESS in 53m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4ec67b1227754382ba93e10469cd9021 : FAILURE in 41m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90c0cc0e9cbc41952e409c1ac2895f68ab5940ab","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-10-24 14:47:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4b0dc1bf9c763dc2e06bfbb49a6418c859064c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-24 15:43:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d06423a991cf277e1fa8c62b0fecd648c3145388","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-24 15:43:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"42246168b44ed2a6ec7949556894ddc25c0fe40c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 17:47:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a5be05d167847b88c96c800b3f5068f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01779723b124e508c9f1b1f798f476f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f83b06ba81c48c4a72d53dd1203d9ff : SUCCESS in 13m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbd40067c3b04579a4c020e11e103bdc : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfe19c76f6f44f6a1de6e650da0ba92 : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4517be0fe97d4cfb8c52ca89dd98c84a : SUCCESS in 12m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/26e8c3e354ea4110967b8946f7ac0707 : FAILURE in 46m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75fe612ae067429882c3ba9641f0793f : FAILURE in 50m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/939331ae4e2f4a4eb04aeefd2cd211de : SUCCESS in 1h 57m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/af783811de7145fba6c7d1480d9a65c1 : SUCCESS in 52m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/232cdb024323433d96950034fa27438b : FAILURE in 50m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/18bb65d3693745349048259eff4923f1 : SUCCESS in 55m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e66bea6bcb15465193f854e0e6a7cf25 : FAILURE in 40m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2baa8fb195da4b968be6d1068ab6fe5d : FAILURE in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a59cbc6b91dd5faa284fb1d9fe1ebf3df93621ba","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-25 10:02:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"109c51a4c2dd45b7c7a5391de983a2e1055ec012","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-10-25 10:30:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"394d7559e29919ac55be64c1d0a8e8e87c34d352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 11:33:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3ef715cf40a43a5aceb7b3f67a6924c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7d3ff13703431aae5bd11bcd7ae49e : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30596dd2dd5b478fb3e31fbf7bdfd97a : SUCCESS in 15m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50e119885a79412388014052fb576036 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94dde1b37a994f6096c5555d16f7797f : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c014a188d6254c71962f9318ee250f45 : FAILURE in 16m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4dd4f8a4318c49c3b3d043128b60cb14 : FAILURE in 52m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee317d236784281bffbd661eff8059a : FAILURE in 44m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c9560be9e7047108929c1827b9b1a1b : SUCCESS in 52m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b51a67aa6b994762b351169ebe893d30 : SUCCESS in 35m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/57da135b7fc74785a88e4d6f7c32fcdd : FAILURE in 30m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/73104f75aa6b4bbb9b6c917a8f28ac93 : SUCCESS in 53m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/52a1075c25e943e3964fb5a64c46f85c : SUCCESS in 56m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bca6504362174de1b0805c46d051712e : FAILURE in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f7a77fc75f34034a2e836e70c4fd26c7e98a7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-10-25 13:02:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c81c64eebd032f4740c03fdc9325046920f4c9d","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-10-25 13:03:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8404baaf908e3c293f0eb01f0182627907565a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-10-25 14:04:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ccfe62a595e6457914d40fa7d38e3051d1b8b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 15:06:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e5b7557f6104e6db936073d6c8080cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24fed734f5104d0d8f145d52bb60349e : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/107e14dea8fd43d3bccca7ef172bfa24 : SUCCESS in 10m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20b684ebd1d5431ba65bf00d29912676 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/822a96f43f7f460889696aa465c76306 : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2de9fea2a6949519ea0830de3feb7bd : SUCCESS in 15m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bd23c56f097b4d8f8d777a56bc057f43 : FAILURE in 29m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21455d5aa5b34b8c9b656a42e0360706 : FAILURE in 29m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fe1a1b5c62a4889ba4c727d12ee947c : SUCCESS in 33m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1dc3430b360d4b1486a672d5e02ff79f : SUCCESS in 33m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/047f942eb78b42ac8da818381a26e180 : FAILURE in 44m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ee5279641dcb458e8d5cfc9be0887461 : SUCCESS in 55m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fcb0ac2ffca245099e69335419488716 : SUCCESS in 37m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a7316bcb30664b88a0e8948709334048 : FAILURE in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"65c344d45be509b8f77719847aba45c824feaeb5","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-10-25 15:58:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3581a93bc3f22a98a4badcd1bdece8bfdab4f907","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-10-25 16:10:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7409a794636a415429b31054e33791d0b3f6fd16","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2023-10-25 16:15:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e1b1ebcb271437c1a286decae9dbb57f90169c6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-25 18:33:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"02886e7b7093ef67183ed8158eabea2edc49ae84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-09 22:32:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8b75c214fb39a788cd6617e6c1b52526dcc18944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:30:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265ddf8202104f339fb6dd8a003bbb6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4fc5eb9152e41d79b3235480e0b8ec0 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/452b3ece21374c5283e4860e47b97133 : SUCCESS in 13m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8b01d1016c9422d88fd4d254259fe2f : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fb2754c98e43b38aa74c77428922c8 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48391b224e49438c92a6b6e1e60b99d5 : SUCCESS in 12m 16s (non-voting)\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13bfa819626e48ca8df74ab96e8b9e22 : SUCCESS in 48m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2bc0bc28304947e092494589a09c21e3 : SUCCESS in 49m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/85852894ed944b3bb828e355462c93eb : SUCCESS in 41m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/05e4f52d47c645ec8fd8f779edf74ea7 : SUCCESS in 39m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/788b1c85b8e44ac0970ecc1fea876af6 : FAILURE in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4eb06813a7944bbebb6958bf9cb1413559e3a024","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-10 10:30:17.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"355e03197f16276cd5160d4a8a00e8aa3b3d8b9e","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-10 10:41:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1e78f7183dd300c1da4300cf9dcff4c3590fa29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:05:21.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"04547ff5bee91d1c3a69c6d335a6d6c4ea419736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 20:58:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2190b8d794645db83721ca10e293fdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a3e66c4da54b77ab8cded156c69d8f : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1219ce9efc34f59ba5ad08c2608cb9f : SUCCESS in 16m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df610b918988498b8b94c2210f8277f3 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5e5b4bc74d4519896a2a1f3e819745 : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2463753548f3490b940ab1c1fd4190c3 : SUCCESS in 16m 14s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/48f46d9754bf4d02985e6ab8c65e2179 : SUCCESS in 44m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c950a9611e934dcab2db38327f20ca54 : SUCCESS in 42m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dec503715c65491c907893716b4cab27 : FAILURE in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dd727587b73f96756f2daa0669a366420fc5a44e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-11 19:33:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"366ce0ed78c4ec747e85cc7ebe26e9882bb0bd44","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-13 05:36:22.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"52a589a0394cd8031cd36a0376bdb5469cdfe94d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-14 07:59:14.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"09789f07a37b84120d3fb18c62619efc876dfe4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:05:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63c9b68e9c18435b836d874637554d1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b3798b14d64a88a51a238419240032 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce269fe9087f4f6ba0f63c053a79c30f : SUCCESS in 13m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29712a7fbd724bbd90bf1734ab462ea9 : RETRY_LIMIT in 1m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5af525e08f448379ec4ac3d6232040d : SUCCESS in 9m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa4ae59b11b44864a114d4d587aeb00d : SUCCESS in 15m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/13f918f5e50b473cac494d4b101383a8 : FAILURE in 36m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7875c9e56a9b4870b855192e383f2e84 : FAILURE in 38m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0f74df8003a4875b3b40df3e49f15b5 : SUCCESS in 38m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/65ce601f423943619026f67d6cdab1c3 : SUCCESS in 45m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9025de49602b457db73cf3b4716d34a6 : FAILURE in 43m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/57108cbddc3e45c889563593edb881d9 : SUCCESS in 46m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8d8aef09cce2480ca4a64ce82a29e3be : SUCCESS in 52m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/484d8adb7c124ab1b5076bcf3a0aa2d2 : FAILURE in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1b5a16a70596b07c60f6db4c841824e9342c09a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:52:00.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"ac2b88bbeba3fdfb63e5048b9ec48bb50b8fbedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 20:52:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b65c968c6b14713856d744c293161ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251178a8098e44c78651c17f5108d940 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42d978af200f495b944a1965902e23c7 : SUCCESS in 11m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11207f6216314020a9efabe5c68d9fad : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21263c3e89e24bfa8636c4419246cc0c : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71401d99613f4627b9e8fd5cfb02da01 : SUCCESS in 14m 35s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c18ec7b4db7a4f9a8b6f89f9b3bb84c7 : FAILURE in 49m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a3152eef857a401d8dd7b22ec31ae222 : SUCCESS in 53m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5951e782926c45baafd12e80dc32640f : FAILURE in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"641c04dc4d761ef8b2343c4304ad729a1d8bac2a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 21:07:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4623706f45c7df69e36622b03f2ad1f73700a88a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 21:58:30.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ead8e5198c074c04ad8848116b0eab24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e022a1cf6f4c52bd897e8c6f44cbd3 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e8f870554cd4a84a92352b0f332d89c : SUCCESS in 13m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3122cf36f194f76a87a8c4def9ef92b : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c7138d44164431b0ac6127129b1783 : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a89c5793fac4467a9781b33835b20d1 : SUCCESS in 14m 17s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/03bccd8d0001457aa04987e84c73622b : FAILURE in 47m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/79c8b85ba4614296bae213235dd64b64 : SUCCESS in 42m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/15d32d8271d345099c495ef4418e6353 : FAILURE in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d08bcad26861e397f958e8aeeb5335e11d6bb46b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-21 14:24:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1d25cebc4552ed145d259de2d3336252a40aaca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 15:25:17.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d80595846a14382ae4a9e6e602c25cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b9e697ca6545ec86ec073894ea075d : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89d173a8be974428873c69c772d6404c : SUCCESS in 12m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8057c66402044a1b51c71c681292631 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730f1c4355d3483e8d9f4b23be1044e9 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f61c9afc9344f2289c093d6bbee8038 : SUCCESS in 18m 23s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7fff9fc779904cecb188b80d6bfd8ae3 : FAILURE in 54m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a67c588e93ed4153947d54780a964c49 : SUCCESS in 47m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/86f807f68aff40878a10b36540264d4f : FAILURE in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c27770881f5580df96ab0794933e3248a950360b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-23 10:45:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2ecbf51ae2c5e1934ee7b3f1687f5e8d6ebee6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 11:47:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c048795d23c41a6a58aa2af08b55158\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe44c2915f14684ae6217ecd437b88f : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e1a6feb26c34b60b22c620dd0e5a07c : SUCCESS in 14m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d89f4e98bb1c4d448432c15af49c445f : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80496d5491ac4d3389566008e1157220 : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e23846a49c6441c99cff250d591b3787 : SUCCESS in 16m 28s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/38076e596912462fafcc9e72a1698fb1 : SUCCESS in 43m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0fb9865289ef43f584154b42c85e16e4 : SUCCESS in 55m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/67c84e16e35f499b9338f822ddca2822 : FAILURE in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6d43b25a03c8f6e2744c6e28ae2724f6fe325ca8","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-23 15:53:46.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4c5cc4e61c450d1d9b2f9b8b5f7b3c1b681afdbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:56:16.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f17b16cde145ef041029f89f14bd1d733d407910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:59:07.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5d246a0ad04cc6a9b9d6a4203cc5b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fddbee9b69744118e5352952eb982b0 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff380283e87344259fda2499766a34f8 : SUCCESS in 15m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6771a6f607c410bbf17e07164cd64c1 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cba36c35d484883a3199555c85362b2 : SUCCESS in 6m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a901a92241fd49a8a7f96fbd7ff0e8f2 : SUCCESS in 47m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/755364cac593403c8ed3bfaee0aeef43 : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":8},{"id":"7cbed6b6b069cd18099748e4627149449eed6695","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:59:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e343ccb078d8ba261ac70afca93f4358589730d3","revisions":{"33f619fa4a225903672556870ec9b43b7d40e937":{"kind":"REWORK","_number":1,"created":"2023-10-24 13:25:08.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/899169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/1"}}},"commit":{"parents":[{"commit":"7baff017d0cf01d19c44055ad201ca59131b9f94","subject":"Merge \"Fix test not waiting for action to be created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7baff017d0cf01d19c44055ad201ca59131b9f94"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33f619fa4a225903672556870ec9b43b7d40e937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33f619fa4a225903672556870ec9b43b7d40e937"}]},"branch":"refs/heads/master"},"5fec682445b5e14224a4f4f191777a53bc6d350c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-24 15:43:29.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/899169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/2"}}},"commit":{"parents":[{"commit":"6480a2c488c568aa2ef13dd7dd89a590170ea1b7","subject":"Fix bindep for Debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480a2c488c568aa2ef13dd7dd89a590170ea1b7"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2023-10-24 15:43:29.000000000","tz":0},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5fec682445b5e14224a4f4f191777a53bc6d350c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5fec682445b5e14224a4f4f191777a53bc6d350c"}]},"branch":"refs/heads/master","description":"Rebase"},"69ea8a3075dff6614f2ae1e0ed818f114f0d36ed":{"kind":"REWORK","_number":3,"created":"2023-10-25 13:02:50.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/69/899169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/3"}}},"commit":{"parents":[{"commit":"6480a2c488c568aa2ef13dd7dd89a590170ea1b7","subject":"Fix bindep for Debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480a2c488c568aa2ef13dd7dd89a590170ea1b7"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-10-25 13:01:45.000000000","tz":180},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nCo-authored-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/69ea8a3075dff6614f2ae1e0ed818f114f0d36ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/69ea8a3075dff6614f2ae1e0ed818f114f0d36ed"}]},"branch":"refs/heads/master"},"fb7a6847320298f79929b97314fd8c64aad7c228":{"kind":"REWORK","_number":4,"created":"2023-10-25 14:04:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/69/899169/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/4"}}},"commit":{"parents":[{"commit":"6480a2c488c568aa2ef13dd7dd89a590170ea1b7","subject":"Fix bindep for Debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6480a2c488c568aa2ef13dd7dd89a590170ea1b7"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-10-25 14:03:50.000000000","tz":180},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fb7a6847320298f79929b97314fd8c64aad7c228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fb7a6847320298f79929b97314fd8c64aad7c228"}]},"branch":"refs/heads/master"},"f25fef9f03079e78869509551a6dc84575e96456":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-09 22:32:54.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/899169/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/5"}}},"commit":{"parents":[{"commit":"81454ec251f053fdf4ccc8615ba374ee4e5bf37d","subject":"Fix bindep for Debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/81454ec251f053fdf4ccc8615ba374ee4e5bf37d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:27:18.000000000","tz":60},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f25fef9f03079e78869509551a6dc84575e96456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f25fef9f03079e78869509551a6dc84575e96456"}]},"branch":"refs/heads/master"},"ae69263f8e910a8d76ac965ec35fc4c0df10dcb2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-10 20:05:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/899169/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/6"}}},"commit":{"parents":[{"commit":"e324c4552daa9b94225e1223e4c008380cc9d470","subject":"Fix bindep for Debian bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e324c4552daa9b94225e1223e4c008380cc9d470"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:02.000000000","tz":60},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae69263f8e910a8d76ac965ec35fc4c0df10dcb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae69263f8e910a8d76ac965ec35fc4c0df10dcb2"}]},"branch":"refs/heads/master"},"364103d1e337074b1d4ae7fcac06cfadc0e13ee3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-14 07:59:14.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/69/899169/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/7"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-14 07:59:14.000000000","tz":0},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/364103d1e337074b1d4ae7fcac06cfadc0e13ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/364103d1e337074b1d4ae7fcac06cfadc0e13ee3"}]},"branch":"refs/heads/master","description":"Rebase"},"e343ccb078d8ba261ac70afca93f4358589730d3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-11-14 19:52:00.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/69/899169/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/69/899169/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/69/899169/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/69/899169/8"}}},"commit":{"parents":[{"commit":"ae0289e029757287fff787ba302fc8cb465bbc95","subject":"Temporarily disable failing jobs and fix bookworm bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae0289e029757287fff787ba302fc8cb465bbc95"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-24 13:14:53.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-14 19:51:51.000000000","tz":60},"subject":"Add api_version for etcd3 coordination URL","message":"Add api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ieade7a30a192910fc3870b173017efd481f0b18f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e343ccb078d8ba261ac70afca93f4358589730d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e343ccb078d8ba261ac70afca93f4358589730d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
