)]}'
{"id":"openstack%2Fmistral~900038","triplet_id":"openstack%2Fmistral~master~I22dd3daae068a0cb287904eb60c6dbf273eaaa10","project":"openstack/mistral","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-09 23:09:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I22dd3daae068a0cb287904eb60c6dbf273eaaa10","subject":"add pyproject.toml to support pip 23.1","status":"ABANDONED","created":"2023-11-03 12:57:26.000000000","updated":"2023-11-09 23:09:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4299dd14de6651408ed9e3e102cbc8da8e0c2524","_number":900038,"virtual_id_number":900038,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-03 13:46:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 13:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"969f41c8d592cfe10b02b59355f6fb5459135ff0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-03 12:57:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9de74b429a2560edfc3c68c145661abb7cc21908","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-03 12:57:43.000000000","message":"Topic set to pip-23.1-support","accounts_in_message":[],"_revision_number":1},{"id":"519fcc88affa6665a0274c30102a1a4375c275b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 13:46:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/970d38098e6147258f90fd93ef6a060a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c5d57883144b60bbff620ba5b6a047 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca23fcca70d54fe699563ec25d40631d : SUCCESS in 16m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/973c22289f344be4bc346e1296b43123 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac24d42591b48a780d3f35d0791010f : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e6c5d6729d14435841cd3d68f2cf7ed : SUCCESS in 20m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a0ad42b3ac6c4fcc983a8c19f7341a1b : FAILURE in 26m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99b8b4b004254310849a39a21c2dae8a : FAILURE in 28m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d7e4eb632d4e0680c9c62ac41c8f2a : SUCCESS in 31m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/201262890cc649dfb637688c0bb6dbac : SUCCESS in 36m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d5051a3db2be4cc8ab13558d68fabe96 : FAILURE in 34m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6483c5abb91c478ebc3a89f8ea6e6bdc : SUCCESS in 45m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/81b04b3f5bfb4e3597fa8c8fe7c9f076 : FAILURE in 45m 34s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4d1da0ed34444d6ba5a1c384e1ec15bd : FAILURE in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4299dd14de6651408ed9e3e102cbc8da8e0c2524","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-09 23:09:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79eb566eb562e99913bc84fbf97c04c2c04980a7","revisions":{"79eb566eb562e99913bc84fbf97c04c2c04980a7":{"kind":"REWORK","_number":1,"created":"2023-11-03 12:57:26.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/38/900038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/38/900038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/900038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/900038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/900038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/38/900038/1"}}},"commit":{"parents":[{"commit":"fb7a6847320298f79929b97314fd8c64aad7c228","subject":"Add api_version for etcd3 coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fb7a6847320298f79929b97314fd8c64aad7c228"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-03 12:54:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-03 12:56:57.000000000","tz":60},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I22dd3daae068a0cb287904eb60c6dbf273eaaa10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79eb566eb562e99913bc84fbf97c04c2c04980a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79eb566eb562e99913bc84fbf97c04c2c04980a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
