)]}'
{"id":"openstack%2Fmistral~900565","triplet_id":"openstack%2Fmistral~master~I6b0896d1805cec9b4cda705569eb57e33208f324","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-14 19:52:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6b0896d1805cec9b4cda705569eb57e33208f324","subject":"Temporarily disable failing jobs","status":"ABANDONED","created":"2023-11-09 22:32:54.000000000","updated":"2023-11-14 19:52:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2014ccacefdedfb9a86a4275f304635657f5beb","_number":900565,"virtual_id_number":900565,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-10 21:06:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 23:25:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"181b2e7fdf5cd3f2ef2867c35e8e82512b7695d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-09 22:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a2d9ebe9b0d1a5959ad0a1b46c05239f56c6b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:25:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72492c807fb44225941544362934b48e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c9e73589d3d4efdb651ad8df371e1f5 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8adb2c923c1245a8a6098631671fcfba : SUCCESS in 12m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4c2985c64ab402b971530da132df6a4 : RETRY_LIMIT in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf52b666c8f4ce4b36b0c26bde5a7a3 : SUCCESS in 6m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7d8be48972429c929ddb6715f47759 : FAILURE in 44m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/631e955a3b3140f7bbe33c4da1052f16 : FAILURE in 37m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/39453ef409404683b7091abae0e04763 : SUCCESS in 40m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a0b4ef8f32cd4f46873cd0b4e83112f5 : SUCCESS in 36m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f5d796a4c97d4328b23124a0f3658545 : FAILURE in 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6294e07cdf6f0e832f4f229f271c115e7b6c3f1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:05:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"448abe341285667442b26e68afa18aa7643d38cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 21:06:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7453f4e9c87844c7a16876531b5ad7df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634a8ba0475b4026acf75d34703876a2 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a13464c7f18540b8a3cec01ea4a7a0ac : SUCCESS in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6addbe22666d4b1c943635d4c620816c : RETRY_LIMIT in 1m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944722ee1fff41c3a26f62e8342d5a23 : SUCCESS in 6m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d49af858dfbb46028c72ad511e078740 : SUCCESS in 54m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0447320ccf42403cab205ed9af3b02e0 : SUCCESS in 41m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/35acfcc2462d4fe5b9a661506e1c9bfa : FAILURE in 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2014ccacefdedfb9a86a4275f304635657f5beb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:52:56.000000000","message":"Abandoned\n\nabandon this one as we are fixing it alongside with 898981","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06f9808da37b3ca9cf551cbd18f03e93127faa18","revisions":{"113ff41dcdfd31c4d63618078ddfcc7d0391a6d6":{"kind":"REWORK","_number":1,"created":"2023-11-09 22:32:54.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/65/900565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/65/900565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/900565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/900565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/900565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/65/900565/1"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:25:12.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:27:18.000000000","tz":60},"subject":"Temporarily disable failing jobs","message":"Temporarily disable failing jobs\n\nUntil we fix the mistral-devstack jobs that rely on wsgi, temporarily\ndisable the failing jobs\n\nChange-Id: I6b0896d1805cec9b4cda705569eb57e33208f324\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/113ff41dcdfd31c4d63618078ddfcc7d0391a6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/113ff41dcdfd31c4d63618078ddfcc7d0391a6d6"}]},"branch":"refs/heads/master"},"06f9808da37b3ca9cf551cbd18f03e93127faa18":{"kind":"REWORK","_number":2,"created":"2023-11-10 20:05:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/65/900565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/65/900565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/900565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/900565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/65/900565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/65/900565/2"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:25:12.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:02.000000000","tz":60},"subject":"Temporarily disable failing jobs","message":"Temporarily disable failing jobs\n\nUntil we fix the mistral-devstack jobs that rely on wsgi, temporarily\ndisable the failing jobs\n\nChange-Id: I6b0896d1805cec9b4cda705569eb57e33208f324\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/06f9808da37b3ca9cf551cbd18f03e93127faa18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/06f9808da37b3ca9cf551cbd18f03e93127faa18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
