)]}'
{"id":"openstack%2Fmistral~900566","triplet_id":"openstack%2Fmistral~master~Ideed8421427e865f8010b7bfca58c3f010ac6bd6","project":"openstack/mistral","branch":"master","topic":"ci_fix","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2023-11-24 13:33:23.000000000","reason":"removed on reply"},"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-11-23 15:53:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-27 17:59:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideed8421427e865f8010b7bfca58c3f010ac6bd6","subject":"Change the wsgi path for devstack","status":"MERGED","created":"2023-11-09 22:32:54.000000000","updated":"2023-11-27 18:04:50.000000000","submitted":"2023-11-27 17:59:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900566-ci_fix","meta_rev_id":"53b11a5dcb6e0d6968ed3d5364c465dae1a2f927","_number":900566,"virtual_id_number":900566,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 17:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-11-24 13:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2023-11-23 15:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2023-11-23 15:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 23:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-10 13:14:15.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"},{"updated":"2023-11-13 08:46:28.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-11-13 08:50:22.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"c83fffd381fddb2799a0c283f6389235d05a4a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-09 22:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fc6ab4328844eed612db5f68c4595b333732e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:43:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b221f051964ed4960908204b3b3de3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df7863186d624fed9c143510bc2218e9 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee0543642fd48c5b02182ea643b5c60 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cee6255e05a747299b3aa912295a1579 : SUCCESS in 12m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e637d76fc55048aabb3ce79b929c737e : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ea6fb0905248e0a751a9ae5c89d4c6 : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d446fcd1ea44446ea2fd9eec01b8f656 : SUCCESS in 12m 11s (non-voting)\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/552493db17d14721977f4742593eae99 : SUCCESS in 1h 04m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a7c6d77ce40f44809b578be0198fe418 : SUCCESS in 46m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7b848ffa1dd747aeaf0ee043b824a5bc : SUCCESS in 55m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d463174696a34fd5929a4c3f1a6b1cc0 : SUCCESS in 40m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/980ca2aef0334e039f7ecf176603842e : FAILURE in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ffc2d5c1653ad50ec6c4462be627ced2c2a56b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 09:21:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e64e22b779bf697aac773aacd68035918f38a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 10:27:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a640e03ed97f46b98ee9a72a5a340dab\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf5d2726dc72402d936fdfad0b8fc512 : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c717d7d6dd424eba7a869d59473008 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb6675bec2dd4569a8d2eacab99c3a8b : SUCCESS in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3128d88ce554ed09ec5e6253287ef6e : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6d372a64194c26b8f61542a0fbed65 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbb4b0d7b4234211bbb111ab6e35f741 : SUCCESS in 11m 58s (non-voting)\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dec22b3c95f941ea8246545efe7273d2 : SUCCESS in 52m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0de9d6d1e8b2465486e31415de6c9dcf : SUCCESS in 32m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/02a4d5e73271455babdc1fdb137b8405 : SUCCESS in 57m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c5015686a164b858f7b83a3ac6addf5 : SUCCESS in 38m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7fbf1b477a2f43ae828cd3094f8964e2 : FAILURE in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"884752c9c108395551f4d0a0e47a3ac2d5cbdb42","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-10 13:14:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6216c7607fd0565ddb8d33bef7a7b99a9942ff70","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 15:47:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06092c406f095473eea926d88958772ce31a0031","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 15:47:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"749fe1711a00ff5cba483e8c7a73c06e5def2026","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-10 15:59:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9edc84e34ff6437377680598c8577a16eb62dd0b","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-10 16:02:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aaeb91e9867e11d6e644ef702dc21518af79e81a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:05:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b55b91c1f1263651362eb98513c0a1cd7c1a747e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:09:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee62387d54e19b5bff91c7b2e5a10aa833f7de12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 22:16:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46837a0e52d5427a8f2c3a041a9fb8bf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e98e7e49dcd3416bbe9ea9bf5283da28 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61bd879ce164bfca5902a101e3442a4 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1581d29c259046e890ea9adb48de0528 : SUCCESS in 16m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06d043bd36f0478e845c6933e7075d57 : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/146b68f511124ff6b17c67c9dcf1dcb4 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7abcfa104be34bc3b72b6c6cdad5984f : SUCCESS in 14m 00s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5c59abb4edca4d12baeda354997ffd98 : SUCCESS in 42m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9084750a167a49f5a2b46640f1362369 : SUCCESS in 51m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/76b1a128a3344dfb9c65ed23535817b8 : TIMED_OUT in 2h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c6951baba5cc64c1a3245bc09d6ed41ed5452ec","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-13 08:46:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1ae0a86cb738240bbd53ab006cd03dc6e7b21679","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-13 08:50:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1f1a6dc6140905e0c8dab2024b467b55b58fc2f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-14 08:01:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a9d3d4136a896e282c16b318af81af04963bd258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 08:51:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d25d572faaa4f1d863a42f3b3d9bdff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/deaf9457e0294cc7b8cfa30839306f3b : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2365b43a8e5443faffe455281d2d7c4 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d1f336e16f4026823320b0964dbead : SUCCESS in 13m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8485f177e3a843059898aa637a9ca324 : RETRY_LIMIT in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19255520931c4640a148894cc17b5508 : SUCCESS in 8m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae54697ea45e4164a9dfacf792be2b3f : SUCCESS in 18m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e93b0093825240b59e714c0206eddfd2 : FAILURE in 20m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12aea387f6a845c28fba0956b9ab6b9d : FAILURE in 25m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d295d4c7a774786b095bb8720a83e2e : FAILURE in 19m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b78db54764784a10b183a71721a7d837 : FAILURE in 38m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1001c5f4f2a543e6813936a428f4ff76 : FAILURE in 37m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cd94bb2f3e6541ccbf572f46abcc9229 : SUCCESS in 39m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f91913662c11402c90f632f7c0123437 : SUCCESS in 41m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c9f9e32c11f24e8ab868c9bacc68a7ec : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb2f0dfde4fdd0d92e04f9bd6a32935ed90843bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-14 08:58:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"276d371ebbe34506a6b91475b347059f95da25f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:07:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18b353bbecc249879631186d032e7f25\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c13b86c7b5744ac9a16f01d73914f79e : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88df9651dc2f4c1ebee20f7e2a3080e7 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d05a43661c26409db057d055d0867e12 : SUCCESS in 17m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c4b09554a5f496c995104cb20a73a98 : RETRY_LIMIT in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e27f34b875422496b743ca03df8922 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3491d4997f4455b7156f1f67bc9500 : SUCCESS in 17m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/30ce07c7180946b5b877f96ae246b32d : SUCCESS in 44m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d48cc3afaba4775a08270a26685604d : SUCCESS in 51m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5956b4319274572a889c78e0f733495 : SUCCESS in 34m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a1e14d31d4f84cc4b63f771df4eea6dc : SUCCESS in 40m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a358e1f6202e471299eade34cad26c0f : SUCCESS in 28m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dabdd476e8444849a8277c4c94f75ec9 : FAILURE in 47m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b8c9b7c0310a4014b47ceb6a418310ea : SUCCESS in 54m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dbd400009c6d4ae1a79a60b62e55d9bc : TIMED_OUT in 2h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b02ed3e4f22f092820346b0bf0b0a529cfe8dcc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-14 11:17:17.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4e3d6bdc846fa550ce5bcaf03f81ccea89bff2d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-14 11:21:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e69812230a6786e80cf2ed2eabb20de74cdc88a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 13:25:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7de88e5cd7ce4cf7b86aa42772d185ac\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d79e60c297ba48cfb898d249e6bced27 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5416321bf7425c8c4f65ad15e951f3 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/787b90b6e6264dfe823933671bde9836 : SUCCESS in 16m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41f8cb1b19654993a9d3946574f350e4 : RETRY_LIMIT in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f46ae4dbc274f138fff3e97d07d07ae : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc1c7a37c4dd49c1b628901fd25faafe : SUCCESS in 18m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2e79dc78482c40dd96326355e1f9bd4b : SUCCESS in 49m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9def3510beb842ae8670ccedd364f86b : SUCCESS in 36m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4218988fe19b451180d2455db0b1bd81 : SUCCESS in 47m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d1d4c3be28494c7e95797925643213a4 : SUCCESS in 46m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/45b4bcb174ad48819eb6dce939ff2c36 : SUCCESS in 35m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8daf4b879d014347a94c2562cbb0502f : SUCCESS in 47m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/691047621afe40aa8a0824bfa33cd84c : SUCCESS in 53m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4cc28ab254d74a03b1fd50c623d31c0c : TIMED_OUT in 2h 02m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4028f91173307cc781cdbe9427a7b4ac218ed064","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-14 14:15:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7b36fc93846d7ceb0e4ab39e2cabc5d8f76b4aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 16:27:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e3a8640a15245548863992380278fea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22d659d73f0c494694b1ecb062cf3d84 : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2b494832cb40d7a6700ddb6a6f4f3b : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/783308defe2045b39e0c58cc0ba67900 : SUCCESS in 11m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d943ada9dc264a6d9c5977239b846e8e : RETRY_LIMIT in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486614dc04984c2b8bbbd68305731399 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/038eb85dc9494408a9a5c05ff56e78e8 : SUCCESS in 13m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/32faa9755d7c49bfbea8bfbbdb892e3a : SUCCESS in 48m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d045212494ca4f24aa63a16d0dd8654c : FAILURE in 2h 02m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b993e8d8dc64bda92af054249da44f2 : SUCCESS in 34m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe794fd249dc40469ebefef5ee882d03 : SUCCESS in 57m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/38e8191e595e40b6bffcb9bfabd221dd : SUCCESS in 47m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cb17982810834733899fd2b3d883e87f : SUCCESS in 41m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3ee09b9c18804623a69a0dac9b4ed7ad : SUCCESS in 39m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/24ab46eafe8744c0ae3242549e2aec10 : TIMED_OUT in 2h 03m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"23ec86d5cd4014ac6a3869febe636f5154804bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:52:00.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"71d23cdb7876c99ad272f8acb15c91975c1b562d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 22:05:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9071f9da6ac454191ebf43107517ae5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6587a61950a0457997024ca41723aaf9 : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518c1b49ede4437b90f5daf9d8230318 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a43d51d14e44de9a2243038294c2f1 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3266bf06b7b492da6ae0088523de03a : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bf938d1dff49f8a93d1ef87eafade5 : SUCCESS in 8m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2be878144fdb4633a6fdbaf772aded1c : SUCCESS in 15m 48s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f6916d3b8c7e4ee6ad923d151b0cd62a : SUCCESS in 54m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/eb03c10733f0411d9e7ab78d8afcbf03 : FAILURE in 41m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/445dc59607b5417b939cd15cbcb34b14 : TIMED_OUT in 2h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c370e6957e1721fc40b0f99c192e09c86cdb56d8","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-21 14:24:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9f565e71b03fc2e43c36a96d16decafd22f51423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 16:28:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da5ca350afc4c23b2e84dd241e91821\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d297bc8f6c5549919e586bd944f2ffbd : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d97b9994dc4a41168da65011cb9c1a2e : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a4ba45788a4422fba4985d115f0c120 : SUCCESS in 13m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe36ea6180f74a6185eec293e9c9cf76 : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9259ff90d7ee476396d84e45c6653be7 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c81b02dcb5e64d1fbc4a4c3bafbb7c59 : SUCCESS in 18m 10s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/65cfce318ce2496493d2a1214fcb9d10 : SUCCESS in 47m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/579ba36b6c5e402ba8685115fb382ef9 : SUCCESS in 53m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/521f3f6890ac451197e39dce814392dd : TIMED_OUT in 2h 02m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ea0e0dcca69e1f3198c0832d1100843eb10749a4","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-23 15:53:51.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d2e78eff358ee26ec912e7eb23e7d2a3d9f24ebc","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-24 13:33:23.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"36a20053c7576b4427b01ae12daa29f65fdbc33b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:56:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0813d2bbc09404592bec0cc8a6931b33350c69db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:59:08.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/630fac999ee74fd8ab58b00ad674bb14\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/25804c45c28942d18f1f2ddf7626471b : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe970d308b4408e91363fb16d7e7633 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ebaba43aed0416f8e6dcb10e0d4e8c6 : SUCCESS in 11m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebbf4fbc1d5b45c8ba55d270315c2278 : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6309132a361a4b1b96f1867f0369889b : SUCCESS in 6m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2febada8678c4800ba9c05621bccc1bc : SUCCESS in 55m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/643c45ee949b42eab852d42b762d0aa9 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":8},{"id":"becd148bd24b3a7ddbf380523bf372cc6df1cf08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 17:59:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"53b11a5dcb6e0d6968ed3d5364c465dae1a2f927","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 18:04:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce35bf035794930972855831ace12f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bba812b4cf741249f9c80ae2cc8747b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25","revisions":{"862acf92f0d37fe58e09dfd6e46604ccdbc03a01":{"kind":"REWORK","_number":1,"created":"2023-11-09 22:32:54.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/66/900566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/1"}}},"commit":{"parents":[{"commit":"f25fef9f03079e78869509551a6dc84575e96456","subject":"Add api_version for etcd3 coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f25fef9f03079e78869509551a6dc84575e96456"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:27:18.000000000","tz":60},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/862acf92f0d37fe58e09dfd6e46604ccdbc03a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/862acf92f0d37fe58e09dfd6e46604ccdbc03a01"}]},"branch":"refs/heads/master"},"a830967329369e1b404e16843d04edf8102adcbb":{"kind":"REWORK","_number":2,"created":"2023-11-10 09:21:26.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/66/900566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/2"}}},"commit":{"parents":[{"commit":"f25fef9f03079e78869509551a6dc84575e96456","subject":"Add api_version for etcd3 coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f25fef9f03079e78869509551a6dc84575e96456"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-10 09:21:13.000000000","tz":60},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a830967329369e1b404e16843d04edf8102adcbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a830967329369e1b404e16843d04edf8102adcbb"}]},"branch":"refs/heads/master"},"d18e172a8639eee282b4ed551b6d85a7297a4025":{"kind":"REWORK","_number":3,"created":"2023-11-10 20:05:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/66/900566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/3"}}},"commit":{"parents":[{"commit":"ae69263f8e910a8d76ac965ec35fc4c0df10dcb2","subject":"Add api_version for etcd3 coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae69263f8e910a8d76ac965ec35fc4c0df10dcb2"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:02.000000000","tz":60},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d18e172a8639eee282b4ed551b6d85a7297a4025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d18e172a8639eee282b4ed551b6d85a7297a4025"}]},"branch":"refs/heads/master"},"cf17219407588ab9f67ac2f6db33b304b6e77232":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-14 08:01:10.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/66/900566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/4"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-14 08:01:10.000000000","tz":0},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cf17219407588ab9f67ac2f6db33b304b6e77232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cf17219407588ab9f67ac2f6db33b304b6e77232"}]},"branch":"refs/heads/master","description":"Rebase"},"cd9c0735ed9a873bdf57e9ac9a38b6c650a9f830":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-14 08:58:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/66/900566/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/5"}}},"commit":{"parents":[{"commit":"364103d1e337074b1d4ae7fcac06cfadc0e13ee3","subject":"Add api_version for etcd3 coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/364103d1e337074b1d4ae7fcac06cfadc0e13ee3"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-14 08:58:23.000000000","tz":0},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd9c0735ed9a873bdf57e9ac9a38b6c650a9f830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cd9c0735ed9a873bdf57e9ac9a38b6c650a9f830"}]},"branch":"refs/heads/master","description":"Rebase"},"79e157e46e883da3baeb1b3f577ee308d17b9b19":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-14 11:17:17.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/66/900566/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/6"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-14 11:17:17.000000000","tz":0},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79e157e46e883da3baeb1b3f577ee308d17b9b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79e157e46e883da3baeb1b3f577ee308d17b9b19"}]},"branch":"refs/heads/master","description":"Rebase"},"00a4efba2d6203c20eb50b6b41560b9e867d03f9":{"kind":"REWORK","_number":7,"created":"2023-11-14 11:21:04.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/66/900566/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/7"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-14 11:19:39.000000000","tz":180},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nAdd api_version for etcd3 coordination URL\n\nSince upgrade of etcd3 (see [1]), the v3aplha endpoint has been\ndeprecated but tooz commit (see [2]) is not yet released.\nUntil then, we should force the api_version\u003dv3 to make sure the\ncoordination between engine is working correctly.\n\nCloses-Bug: #2040048\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/00a4efba2d6203c20eb50b6b41560b9e867d03f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/00a4efba2d6203c20eb50b6b41560b9e867d03f9"}]},"branch":"refs/heads/master"},"ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2023-11-14 19:52:00.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/66/900566/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/900566/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/900566/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/900566/8"}}},"commit":{"parents":[{"commit":"e343ccb078d8ba261ac70afca93f4358589730d3","subject":"Add api_version for etcd3 coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e343ccb078d8ba261ac70afca93f4358589730d3"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:15:17.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-14 19:51:51.000000000","tz":60},"subject":"Change the wsgi path for devstack","message":"Change the wsgi path for devstack\n\nSince pip 23.1, the wsgi_script from PBR are not installed anymore\n(see [1])\n\nSo, instead, rely on the wsgi.py app which is directly bundled in\nmistral source code.\n\n[1] https://discuss.python.org/t/adding-support-for-wsgi-scripts-entrypoint/30905/11\n\nChange-Id: Ideed8421427e865f8010b7bfca58c3f010ac6bd6\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
