)]}'
{"id":"openstack%2Fmistral~900567","triplet_id":"openstack%2Fmistral~master~Ia97057be1d15913958f151b2007a2720ba78fc5f","project":"openstack/mistral","branch":"master","topic":"temp_branch_2","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2023-11-28 13:21:10.000000000","reason":"Change was submitted"},"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2023-11-28 09:32:47.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2023-11-28 13:21:10.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-28 13:21:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia97057be1d15913958f151b2007a2720ba78fc5f","subject":"Revert \"Temporarily disable failing jobs\"","status":"MERGED","created":"2023-11-09 22:32:54.000000000","updated":"2023-11-28 13:26:58.000000000","submitted":"2023-11-28 13:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900567-temp_branch_2","meta_rev_id":"049194cc34dd249954d1f0d563bbdeb6d766d8dd","_number":900567,"virtual_id_number":900567,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 13:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2023-11-28 12:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2023-11-28 12:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 23:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-11 19:32:40.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2023-11-13 05:36:56.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-24 13:33:11.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"78861ad38762b570c9f806364332f81e44c6e950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-09 22:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d6c66aa93e2ab82a9ec9ab991067732c0289fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:36:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/505f38795abe47e8a363e36e70ef1b75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2449d29921284290acae7aa9a3064758 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f36f06376a4c5caf56fb210e526575 : SUCCESS in 14m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59b30b6adbf84d8ba75051b89e08288e : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b91b8a6846548bc89cafd1b35043bff : SUCCESS in 6m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6608466b868b46dfbe976e6b0871a77e : FAILURE in 26m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4f8f3f776224d39ac2fa204aa644177 : FAILURE in 31m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9a5ef71b4694757b7a72765c88f614f : SUCCESS in 47m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7c9407a56d184c85a76a27a3f884e285 : SUCCESS in 53m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b2a8dfe33c7742fca266040ca01f8d9b : FAILURE in 43m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/901d041c83b34ff8a33757b3b10ee9fa : SUCCESS in 40m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5e9f909df16c49d88d994ddd38fe7386 : SUCCESS in 39m 35s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7b4eae611c4e47ae99f0ba0071b387fa : FAILURE in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2700b23ffa3270e4b49035112d0d84a3b153ba77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 09:21:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7676b023c69354d4bf918aeb0374b6eb9387b74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 10:26:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea2cc0dbeeb45f681c2ea7d51549777\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a06d141bfdf45d7baaef2def1770866 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/042ae38ccdc04c6cacef343f28bbf1c4 : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea54fd230d3f4836853d83d3883b6e44 : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfab96e6389d4d9ea627dadf4b084edf : SUCCESS in 6m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/58cb4bef5dc34a55ab9de3900458d3d4 : FAILURE in 32m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9a664748ad24e43ba6cb59952722140 : FAILURE in 43m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4cbe1b3d69349d8a20c570ae3179bca : SUCCESS in 57m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f021baf91a964b0bbb07a4ddf39f843e : SUCCESS in 46m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/49d7e72028b04aafb01d74e6e7bebf80 : FAILURE in 48m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2a056aba30e84d52a76eaa435fd56e00 : SUCCESS in 46m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e51b592849b24c038dcff8321b71e368 : SUCCESS in 44m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1a73b4d3c7d54ba7b5a88db11464874b : FAILURE in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18c43c01f95fa03e44992dfd5b0dfddc63cb1a5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:05:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"020961ddca06195a3597df9ef8b19e55968c7ef4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 20:11:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27bbd159bfc732960d83d6fea63b0a3a5d79a2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 22:13:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d5722b67f9349c29a47505c3a83a794\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464dc968e52545068fa1ba02f935c2b3 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e02aa32fbd594b91b496a4c987e9aad0 : SUCCESS in 11m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/744fcdae80b74f8da1fc5d8826d94839 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26fcb2d4d3f540709787dab15bfc6aa1 : SUCCESS in 6m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f6d6bfad95a34051b202925397066329 : SUCCESS in 48m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/290a96e325ef418c9b7f0a93e7370e1e : SUCCESS in 48m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c30c4f8111234a00b4d43b309be0ecb2 : SUCCESS in 31m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e372ad55f8234b8080db56791d0419dc : SUCCESS in 35m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/097ade1b01e546be8348ceb51666fafe : SUCCESS in 35m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/407809f50c4243e78d94bb767fad1987 : FAILURE in 46m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0e56d6eb8ac14106bbb19184e58d8ac8 : SUCCESS in 41m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d9fcbbc7e4a7470e8de7fed8e5460e30 : TIMED_OUT in 2h 03m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5cc88a969a12ab7796612335ccca9e853ea0636d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 22:45:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"298a8478a8e26b89ab1edd2e473265e05d507807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 00:49:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebe3fb84d5240c39f71bfcdec187c17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a264cadfcaf44715aadb329b6db1debc : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57df141cf1114269bafb80b4b565cab0 : SUCCESS in 10m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a0d732218b7488a9ae96ac230ad4158 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5e0e8fe63f4ba4be82de04e299dbcc : SUCCESS in 5m 39s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1af560f5c47041fea8fa24bb033f9a70 : SUCCESS in 1h 16m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa5f392348ba40fc8becb05e5467f493 : SUCCESS in 27m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3f4c94344cc409db174c5d9359d53cf : SUCCESS in 46m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e64c9aef8c6f41c4b8cea38096492353 : SUCCESS in 26m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7f0bec76b95d4c11bfb105831685c0b8 : SUCCESS in 47m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/24b460e41bf14e32893d1c359cfc56c4 : SUCCESS in 54m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9abdd791ecf947169101a6fbf28a2f87 : SUCCESS in 36m 06s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/30e441f00cc2400ab5185eefe90538dc : TIMED_OUT in 2h 02m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"640f0eb7cfa92ec6928f3e4a8aaf018882cf5de9","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-11 19:32:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b29bca65f0ac74d2d097f218e02f7fda90bd79fa","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-13 05:36:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f206fa3aa91f891e3db6e67f80baaa1f5a5b2298","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-14 19:52:00.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c3ce7a4f4d2a0b8f206920b095b42f1bfbc3cc3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 22:01:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e19dc2e9d4bf4dddb47a368a3b63951e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725816fc7db449aaa0c6049f8017ad55 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f1f8d3851484d56b4d04ecb463a4130 : SUCCESS in 15m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbb6a5d6567c4de9add793fa9c7470f2 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7b3ee3165148708da6403264fb272f : SUCCESS in 7m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1cdf8792621c4a389fe67ac39105866a : SUCCESS in 36m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1bb01a4810a442cbd5ff07d7a729039 : SUCCESS in 39m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81fb0a49697b415c85dfd5c53da2fc22 : SUCCESS in 40m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5ba544d512e649cd96ba5c414ac31a94 : SUCCESS in 30m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dbda188dd43144eea974ca50005763a6 : SUCCESS in 1h 21m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8bca03a9915c4c68a8db9d52742c11ed : FAILURE in 49m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dd92552cda3c422683c078bfb3ea22c9 : FAILURE in 39m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5590fe4449f8441fb1e9b7375afd60ec : TIMED_OUT in 2h 02m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f849fcce8f289812b52e4590a652212d63c14fc","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-21 14:23:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"96c0ecc99af6bceedf6f63fc2fa25bbd4c0684c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 16:28:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85041398f99a401bb026665d271abb40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e539f85b7543dc8a41522b1237f9c6 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaff882c88db493b894c8070f7ce8c04 : SUCCESS in 12m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37d8c38bb8d84691ad0e0dcc7e83e1ea : SUCCESS in 14m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc06199ca0b4e8ba679fb4a9e03dee2 : SUCCESS in 6m 39s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8048da9b68a24ef38e3c762cb163a57e : SUCCESS in 51m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a257a8b037e44368884000411c4cc65 : SUCCESS in 52m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f008d2bc6208444588791a7b50479491 : SUCCESS in 54m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/706062a88a7e401ea5f2c00468f9c302 : SUCCESS in 39m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/85f7128715bb4b0780ac0ead7a942e13 : SUCCESS in 35m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/853b59e177c446559b417fb1452cde01 : TIMED_OUT in 1h 00m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7456e7fac89a49aeade266a510ea2bce : SUCCESS in 54m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/289331455fa344feb5557522aab8e4d9 : TIMED_OUT in 2h 02m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51cb2e281fb11763a53740178611ae7fd844c396","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-23 10:46:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b65086f792586087ccae963be23965e66ff57dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 12:58:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a9caa7c77b4f7cb4770a94d5755b7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40bde68d6f2949518c6a4bb9eb060bb5 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bbf5c8cb34f4eda94c609559f823c6c : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9a3b5d783ef4e10937e69f7b4971305 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57575ef2845a41ec8ea7911455882d54 : SUCCESS in 6m 47s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/41573f99b49d46e7aa2180e0e69bf96c : SUCCESS in 52m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a0cdfde795343138d453ea40f078b93 : SUCCESS in 53m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/026c4598d3d3460794c808cb707909e3 : SUCCESS in 33m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ab883dff30b34222853b325dcfb28230 : SUCCESS in 48m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc6e3cd77b514d34b6205ac9630dd16c : SUCCESS in 48m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/14fa20c976b843199a431be45b7d43c3 : SUCCESS in 54m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a12458f4f9ab49099aca991d20e103b5 : SUCCESS in 43m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/afaf8bdbb022481c9f4505f6127dd529 : TIMED_OUT in 2h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c4071de392797b834c2b1eecd52b5e7fcf4ac60e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-23 15:53:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f8dd25a5a99b28d213d590dd93aecbc909d2d9b0","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-24 13:33:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"396d390d26e130efba5fb1eec1fa22b89b2d475d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-27 11:19:14.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"57e11a745a1090ef6d218dbbac4f38fe84d2aed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:06:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d42313c37812489d871d7cd8585af551\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c5347806676649e49ef9c3d75e690c29 : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae38e36026504f5e90e8a804fd74fcd3 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1600cfed29b458c9a0d26ba7d960147 : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b245ae65c0bc4162ac7064b86f16a48e : RETRY_LIMIT in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89fb94b39934d8d93c0644093b1c420 : SUCCESS in 7m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0ba8149df0694d469f863461cd646560 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2127b7dc4ba47b5b9b8ba1f07983e2e : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9042958df5714bdba9a57ce258477b36 : SUCCESS in 23m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5aaf2bc5581b4b939e0a4bad3a314ff5 : FAILURE in 24m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ef2098e995444cf8b148f9c54e79f53 : FAILURE in 38m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2fdcb3af4b34ae587f1397f622c544e : FAILURE in 35m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e606907f0393418ca40571a4532ef137 : FAILURE in 40m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6b5ac69d6ee145da85523f4b4456f748 : FAILURE in 26m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/abf724aceb8c4c199d33a1ae28b8eb64 : SUCCESS in 43m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e036ad4a56e245fca5b4d2e7d675e642 : SUCCESS in 36m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9e80692709e8416ba6d188685e33f60d : FAILURE in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"164f33aaaf3c4793df7ff76955f6b53eefadf898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-27 19:10:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a59c010e473fa25f3c43d797d8c251430dd405bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 21:14:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3779a51064a1459cb47ad33f67c8bcbd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7084bd2e726a4f5c80b357ac1b239531 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f62aed44cf344c2a38090ad5842856f : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22d2cc67a4d43ab8c060545c345dd49 : SUCCESS in 14m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0addf5e9d2c94391a50942356d2e68f3 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b1124ae91a5438fab011440a5d37171 : SUCCESS in 6m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56251ce2ed0e4d4caaa503a22d170838 : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbe72971863042278fbaff144fed10c6 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad6aeb05f554edb8214010ddcbb2e72 : SUCCESS in 13m 57s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5bee40f6d3844214bf7952ac5f46d54e : SUCCESS in 40m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2e5d991987294d478b4e6fc47589207f : POST_FAILURE in 56m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7216ae66bf234a3b9b5812696226b09b : TIMED_OUT in 2h 02m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"abb3f53dc4e966f0c67bb6756bf0f313f8863ac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-28 07:19:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6fa146f06ac09ce1f7320f2730b87d87ff592a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 09:30:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25ab99c8a8f14ff3be4748629e3f1078\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b82f768d974f7d997cb7bd9f9a9c02 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de17d56e45c14be3835d2d52b31344ff : SUCCESS in 12m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b53032ca4794a259fcf25f301bc94eb : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba81a1c331a4a67b1dc444b3aabbc02 : SUCCESS in 7m 45s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e180a15be2f842259a1081a7164b49fc : SUCCESS in 53m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a0ab44f9cab4218ba547210ea1a4626 : SUCCESS in 34m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b84ff9b196b44c2c9d28925174728450 : SUCCESS in 47m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/368529e09708498ea97338dd2aa6e0fe : SUCCESS in 50m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/deb8b013b7614321a385e40020f88dda : SUCCESS in 32m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2a17867a3cbf4205b78613b7491cbf53 : FAILURE in 41m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b8a87895dbff428999ce1a89bb8ea858 : SUCCESS in 54m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c1dec22962cd4b979fa677adc227b88f : TIMED_OUT in 2h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"06b0aa6b563917fe496bac9e70227c621bbe355f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-28 09:32:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f9fa2a132582d37ddd8c8c92e5a082bd4eafa5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:42:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9dc7177f1b94483b4976bbbc6684e25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4056ea121c1489b8f71faa9635886c9 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53907c5ce692469aa56428f5d2d38b82 : SUCCESS in 12m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e0266fc545c4138949bcc1320f95a90 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5186266553af4eaa836276a6deee8384 : SUCCESS in 6m 39s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fa1f5cc168da4df19d219031f90a8b2d : SUCCESS in 50m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c76e4c121a945779f15bb59668937f1 : SUCCESS in 46m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ecf9db50c5b47ac999118ad61abddac : SUCCESS in 42m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a566d65cc18541dfa1efd590353c6ad6 : SUCCESS in 32m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/10837789c8034348a7c19b67c6672d3b : SUCCESS in 47m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fc0d593c63d6445dbfe38fb4971948ed : SUCCESS in 47m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6fdb538d5a6e42c590ef17cb467944f0 : SUCCESS in 51m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4d7cffc95a344f93880c37888525646d : TIMED_OUT in 2h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8424d92b03fa7ffd494fba2907590fa80c1a8c65","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-28 12:14:20.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6549814438713ef920aa789ceb91e34877ce4e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 12:15:02.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"47a0209c0fbd965564a0595dfab2b5d486d026b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:21:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebd272429a584fd384d7ba6c4b629f59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d43bbc92c924c738a6db66b7b1914e9 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47e2028af9d64c238c542772493b7255 : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05757aa86cec4568857567a67397c27d : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0bfdb029064d658b24f16f07ef8d32 : SUCCESS in 6m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b1f369ae73d74870a7c96a6ea6d0e92b : SUCCESS in 58m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/354ad114c58a487a8f6766dbc435045d : SUCCESS in 50m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ccd2adfd9d3474fb39cb3672e3622a3 : SUCCESS in 50m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a701bf785bc0430d8c32de151322ec5c : SUCCESS in 47m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9501f02595a742c781fbbed09ce5ca26 : SUCCESS in 34m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/234001a7c1bc4eecb3dbefeafd753807 : SUCCESS in 53m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b16d1abb8de34c49acb1ace15ccaa600 : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":7},{"id":"2b2cfb02478785ed3b6fd9c5ce0614af703382dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:21:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"049194cc34dd249954d1f0d563bbdeb6d766d8dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:26:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78ed517c43942a4b352f18e69711350\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c85ad9aecd04e078ccc64da91e8fa1e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5be3de2fd4c1abccbecd768fd2d74061e9d14c4d","revisions":{"0b088e02726afe0d1747851dcd98a5c607fabaa9":{"kind":"REWORK","_number":1,"created":"2023-11-09 22:32:54.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/67/900567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/1"}}},"commit":{"parents":[{"commit":"862acf92f0d37fe58e09dfd6e46604ccdbc03a01","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/862acf92f0d37fe58e09dfd6e46604ccdbc03a01"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:26:23.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:32:42.000000000","tz":60},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit d67cfb24f12756867ec0b3d578f90a0f98775780.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b088e02726afe0d1747851dcd98a5c607fabaa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b088e02726afe0d1747851dcd98a5c607fabaa9"}]},"branch":"refs/heads/master"},"af55ec57778896a9f7a867b7514cff2049d94809":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-10 09:21:26.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/67/900567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/2"}}},"commit":{"parents":[{"commit":"a830967329369e1b404e16843d04edf8102adcbb","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a830967329369e1b404e16843d04edf8102adcbb"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-09 22:26:23.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-10 09:21:13.000000000","tz":60},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit d67cfb24f12756867ec0b3d578f90a0f98775780.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af55ec57778896a9f7a867b7514cff2049d94809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af55ec57778896a9f7a867b7514cff2049d94809"}]},"branch":"refs/heads/master"},"f7d35ee2dcc2e7a23147623b5e6836f135c31e15":{"kind":"REWORK","_number":3,"created":"2023-11-10 20:05:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/67/900567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/3"}}},"commit":{"parents":[{"commit":"d18e172a8639eee282b4ed551b6d85a7297a4025","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d18e172a8639eee282b4ed551b6d85a7297a4025"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:11.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:39.000000000","tz":60},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit 06f9808da37b3ca9cf551cbd18f03e93127faa18.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7d35ee2dcc2e7a23147623b5e6836f135c31e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7d35ee2dcc2e7a23147623b5e6836f135c31e15"}]},"branch":"refs/heads/master"},"fcc5058d5df425576811a295941712812f55366c":{"kind":"NO_CHANGE","_number":4,"created":"2023-11-14 19:52:00.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/67/900567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/4"}}},"commit":{"parents":[{"commit":"ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:11.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-14 19:51:51.000000000","tz":60},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit 06f9808da37b3ca9cf551cbd18f03e93127faa18.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcc5058d5df425576811a295941712812f55366c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fcc5058d5df425576811a295941712812f55366c"}]},"branch":"refs/heads/master"},"3c59968fd1d357dd3eedd778aee3e9fa96698bb4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-27 11:19:14.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/67/900567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/5"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:11.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-27 11:19:14.000000000","tz":0},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit 06f9808da37b3ca9cf551cbd18f03e93127faa18.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c59968fd1d357dd3eedd778aee3e9fa96698bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3c59968fd1d357dd3eedd778aee3e9fa96698bb4"}]},"branch":"refs/heads/master","description":"Rebase"},"319cad0b6c43ded26973a53d14fa711e7a60c9f2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-27 19:10:06.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/67/900567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/6"}}},"commit":{"parents":[{"commit":"ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:11.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-27 19:10:06.000000000","tz":0},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit 06f9808da37b3ca9cf551cbd18f03e93127faa18.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/319cad0b6c43ded26973a53d14fa711e7a60c9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/319cad0b6c43ded26973a53d14fa711e7a60c9f2"}]},"branch":"refs/heads/master","description":"Rebase"},"5be3de2fd4c1abccbecd768fd2d74061e9d14c4d":{"kind":"REWORK","_number":7,"created":"2023-11-28 07:19:52.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/67/900567/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/900567/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/900567/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/900567/7"}}},"commit":{"parents":[{"commit":"ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2023-11-10 20:03:11.000000000","tz":60},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-28 07:19:43.000000000","tz":180},"subject":"Revert \"Temporarily disable failing jobs\"","message":"Revert \"Temporarily disable failing jobs\"\n\nThis reverts commit 06f9808da37b3ca9cf551cbd18f03e93127faa18.\n\nChange-Id: Ia97057be1d15913958f151b2007a2720ba78fc5f\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5be3de2fd4c1abccbecd768fd2d74061e9d14c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5be3de2fd4c1abccbecd768fd2d74061e9d14c4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
