)]}'
{"id":"openstack%2Fmistral~900580","triplet_id":"openstack%2Fmistral~master~Ib6ab36f93b408fdae59c1c08146f2ca069ed4f96","project":"openstack/mistral","branch":"master","topic":"sys-modules-iter","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-13 22:58:48.000000000","reason":"Change was submitted"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2023-12-13 22:58:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6ab36f93b408fdae59c1c08146f2ca069ed4f96","subject":"Avoid RuntimeError caused by iteration over sys.modules","status":"MERGED","created":"2023-11-10 09:28:32.000000000","updated":"2023-12-13 22:59:48.000000000","submitted":"2023-12-13 22:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900580-sys-modules-iter","meta_rev_id":"4839cb5ba16bde4023f917f7086e070484e75994","_number":900580,"virtual_id_number":900580,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-13 22:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-11-28 01:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},{"value":2,"date":"2023-12-13 22:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},{"value":1,"date":"2023-12-13 22:07:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 09:31:49.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2023-11-10 10:52:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-10 13:15:09.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-11-10 15:46:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-12-13 22:07:37.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"250bd78e8de7d8c107c720472427fdae2cc35b2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 09:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0141d8218e4e71121783b0346c9e6ed0f6fc46e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 09:29:49.000000000","message":"Topic set to sys-modules-iter","accounts_in_message":[],"_revision_number":1},{"id":"a6d020fa80d7b9768f048a327717f9fafa2d5c25","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-10 09:31:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bafad79f3662fbfbf4b34091d6ce0613eba4cc3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 09:52:59.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bceaf4831f79cf8b796d4f4679df7a019e23f4b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 09:54:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4a68d7b8bfbc14f729ed03e7a5d6d2d0ba66731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 10:52:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b8820b7721545e7b59a9c868dbd0fa2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3895e70f6bb94038b2c446cc4cd1bf70 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba4cd7ef9b1548a28118ae96a341f562 : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27693cc7cdb744a2a4b7382e32dd940a : RETRY_LIMIT in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77c8eaafc03408ebc6a2a38b3dd9798 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fbccf48cce84148b28c1bbd7d43d7a2 : SUCCESS in 15m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f9af605d094c4a8687e2807afb400c99 : FAILURE in 23m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c26c7b1ac9174ad2ac83044f00ceca1d : FAILURE in 38m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e428a3990ac46ebbb28043db9f682ab : FAILURE in 42m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/987a1ce14b34480d8bec7d5b16808cf6 : FAILURE in 27m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3bff5406d24a413d847be2567021b0e5 : FAILURE in 38m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6567617c652d4364a5fbc9f25f984341 : SUCCESS in 58m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6022dd58fc904b36a378077943e25547 : SUCCESS in 42m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fd627d03cf604eaab4c344becbcbcc6f : FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5dce84b7a622cbe855cf5b1beaa88894348f6600","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-10 13:15:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ceb909f1fb2d2cd477c21bedb254b3fe83d8120a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-10 15:46:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c23fd4298b6b9930cbd181062f8383d10aaebd2","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-11-13 13:08:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d81630e34dd038a8406cd05635064371415744b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 14:08:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09644cbbcb304cf28f9f103294ab97e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe3945494b14262977b7d58ef8e8945 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af305a2b56814c06afb1c6843be49188 : SUCCESS in 12m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4926d8b0afb4e08b80bf53102b2abcc : RETRY_LIMIT in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d716e1de2ae9480e9cdcd6fa038e774c : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef514875e4e64dbe91801a60e134609f : SUCCESS in 15m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/58b614ebe8cd401b8e2225802df27e15 : FAILURE in 36m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4461898ac6f432cba349aaae70d2822 : FAILURE in 25m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/815f8e45f2a24b38be46147e59537ffb : FAILURE in 31m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/956e35773150433495f512cd995323f7 : FAILURE in 37m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7eca8fd1b3924ce98b972f84b5adb818 : FAILURE in 25m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5890ae17bbae40ebad47b99fdfbfa0a3 : SUCCESS in 54m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/600ec29c98b64026bd74f24798e05e21 : SUCCESS in 40m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bd5694e91e2d41ffbab427dccab86efa : FAILURE in 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45ddeaad713caa5a4437ca2cf82c66a5261841c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 01:12:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3b915cb1cb361d26598fdd58baa2993e42635f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 03:21:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b482df98544efd8b7954cf5e6f09c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74cc91f033bb4e3395ff55f209e26841 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb2cb460f53f492e942151ce170e5f48 : SUCCESS in 13m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b67183b423b46f6a724301820f7953f : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f059af6cd67c4ba78dd81c6e94d66ef5 : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4816ffe0f8084300a6c41735a7404af2 : SUCCESS in 13m 00s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2b05195a13264fecb5565074962c4d77 : SUCCESS in 1h 00m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/501fbae087df4ab4abf207a04919fa7a : SUCCESS in 53m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8733faf5360849ef8acd557726430826 : TIMED_OUT in 2h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f2257de56993cb0c8a9cb415cc6f62fb51fe0a6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-13 22:07:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"604e8d457a3bc55defa40c5bed690cf28a138d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:08:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c010fa5dd614c0081d25d7f0b55aa04c59c59c00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:58:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/475634babca34346bb3019c39318135a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f983439879747e8a3734ba6cf32dbfe : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4c62e5866074606b89ddaf18f58c272 : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e8d092e35274de4ba2c0cda73af9f3d : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ed246e9f1645b1a4d27277959a3201 : SUCCESS in 6m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/87b60e8332664c9494acac789f453a83 : SUCCESS in 37m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/357674a477ab48e5aa46bb8c23933ed7 : SUCCESS in 30m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20057aa0b7734ecbb81de7ec24b5bc98 : SUCCESS in 29m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7f6c156cf0cb473e9aeca572aec238db : SUCCESS in 39m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/51624cc781bf40cb85a62404004411e4 : SUCCESS in 36m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/97da2a372efc4c47ab0ddfa688e60ae0 : SUCCESS in 49m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/240fe0f2164a48f789b91bded8fb43c6 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":3},{"id":"24858c330ecd3ee7d846701a58a2d73a566eea34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:58:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4839cb5ba16bde4023f917f7086e070484e75994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 22:59:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08bac5005d043c7a9677d5ddf5f28fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee06602fe854a9480ab4e1d077d7824 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6410ca6518fc21fc36df45108f91c540629d65a3","revisions":{"e0124e1e3d9839e325cc1fc672d5c53988401015":{"kind":"REWORK","_number":1,"created":"2023-11-10 09:28:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/900580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/900580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/900580/1"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 09:26:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 09:28:29.000000000","tz":540},"subject":"Avoid RuntimeError caused by iteration over sys.modules","message":"Avoid RuntimeError caused by iteration over sys.modules\n\nAs is described in the doc[1], we should not iterate over sys.modules\ndirectly because sys.modules can be changed during iterations.\n\n[1] https://docs.python.org/3/library/sys.html#sys.modules\n\nChange-Id: Ib6ab36f93b408fdae59c1c08146f2ca069ed4f96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0124e1e3d9839e325cc1fc672d5c53988401015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0124e1e3d9839e325cc1fc672d5c53988401015"}]},"branch":"refs/heads/master"},"041f08bcaac76020a731f1d390ae73c1a7e9e998":{"kind":"REWORK","_number":2,"created":"2023-11-10 09:52:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/900580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/900580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/900580/2"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 09:26:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 09:52:59.000000000","tz":0},"subject":"Avoid RuntimeError caused by iteration over sys.modules","message":"Avoid RuntimeError caused by iteration over sys.modules\n\nAs is described in the doc[1], we should not iterate over sys.modules\ndirectly because sys.modules can be changed during iterations.\n\n[1] https://docs.python.org/3/library/sys.html#sys.modules\n\nChange-Id: Ib6ab36f93b408fdae59c1c08146f2ca069ed4f96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/041f08bcaac76020a731f1d390ae73c1a7e9e998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/041f08bcaac76020a731f1d390ae73c1a7e9e998"}]},"branch":"refs/heads/master"},"6410ca6518fc21fc36df45108f91c540629d65a3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-28 01:12:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/900580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/80/900580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/80/900580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/80/900580/3"}}},"commit":{"parents":[{"commit":"ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25","subject":"Change the wsgi path for devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdeadeb7a1aa87cba2cdb0c1a2bb1ffc4aabf25"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 09:26:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 01:12:35.000000000","tz":0},"subject":"Avoid RuntimeError caused by iteration over sys.modules","message":"Avoid RuntimeError caused by iteration over sys.modules\n\nAs is described in the doc[1], we should not iterate over sys.modules\ndirectly because sys.modules can be changed during iterations.\n\n[1] https://docs.python.org/3/library/sys.html#sys.modules\n\nChange-Id: Ib6ab36f93b408fdae59c1c08146f2ca069ed4f96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6410ca6518fc21fc36df45108f91c540629d65a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6410ca6518fc21fc36df45108f91c540629d65a3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
