)]}'
{"id":"openstack%2Fmistral~902141","triplet_id":"openstack%2Fmistral~master~I1bb527fe518ecc5548e86ea339fa24e3a76b7e78","project":"openstack/mistral","branch":"master","topic":"fix_broken_tests","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2025-09-23 06:01:40.000000000","reason":"Change was abandoned"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-02-15 22:02:55.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I1bb527fe518ecc5548e86ea339fa24e3a76b7e78","subject":"Fix broken tests","status":"ABANDONED","created":"2023-11-29 13:12:39.000000000","updated":"2025-09-23 06:01:40.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"369cccffcea08aa0139e2669c0424566b2fe7164","_number":902141,"virtual_id_number":902141,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-15 20:34:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-02-15 22:02:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 15:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-15 20:38:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-15 20:38:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-02-15 22:03:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"}],"messages":[{"id":"10844da974cea50ed5e2a1d626746f2a8d86d62b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-29 13:12:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f84ddecd1c01b4aa3e71a7982425ccbe00b7c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:21:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56af2157bb644cb986740b7ee87677c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac6cd9c440841b9af7e7af0dfa83dd9 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d9091323cb4b4892d05dd87180ff81 : SUCCESS in 18m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d97c4ced6e1b4a98b33c560e73fa5f0a : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6600e76f146346218866d848fdebe399 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce2be1e79fcc460dbbaca75e6fa2a238 : SUCCESS in 19m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/28c5c028b7f0408bb5a885c6b2a44201 : SUCCESS in 38m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f623c7e4fbb4e50a23ca21a8c96f670 : SUCCESS in 37m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7662df8aa4ac40edb1ade2dff1d899bb : SUCCESS in 47m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7c73fe45316f465fbfb9c349eb6f3de6 : SUCCESS in 48m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/86f305049e7f4c8a9e9cf00f70c7401d : SUCCESS in 33m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0bc87461b97b46cf9ade25da49595b3e : SUCCESS in 45m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7c5b745101564e38b877df2972a6598a : SUCCESS in 54m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bc3afdb1b4db46fdb2b400a9b008d029 : TIMED_OUT in 2h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b66c26941872f308fd0c46573d36aec9efa3d1d9","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-29 18:39:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70b706c02638751d9c6c8d94e2284080b34f5717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 20:47:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bbde2757bd74a3db0f3fb11a0ead73e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ef504470e0464faf8fed85e70ee5eb : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e8d7a6125cf485b8c3500ca216bf0c7 : SUCCESS in 13m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b0db82285f04832bd32ec5da6b2befa : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f7f72606674389854fa20d57eeabba : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e543536cc2ab4bf88c98e7ca93af70e1 : FAILURE in 13m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7189a5b6fdc5455db33ef70684c8d5c9 : SUCCESS in 52m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/893feaf886c54911bf78c442cb89a0db : SUCCESS in 33m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96045df2ece7487dab36f57ec0f0e84d : SUCCESS in 47m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/05f98e0a109f40e799541bb3f038f4b5 : SUCCESS in 51m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3be5f66b9c0b41d987f55261cf91c5de : SUCCESS in 53m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/96f7b11e09c643d084fcdb1999609881 : SUCCESS in 55m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fdd40718f45b4d16a54c754914e30fa1 : SUCCESS in 51m 39s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1431dd6f0b234afb81b2efc9c7ed85fc : TIMED_OUT in 2h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3ffa5b13ae067ed603f6b0bca2186b7e3cdf76b","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-30 09:27:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b56c81ea6282379c23ea83aee9228a0f77c1c450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a2737ba202245e7a987eca823ed4429\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248d7f8003894c41bf7bcc23f385eea5 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1731069670b404499b653c478f3b1c0 : SUCCESS in 14m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/522835082ba1465e8af63d98cde9acb8 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6dcacdfd6144cf83517beec59e23be : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2698dbffb98d4e20b1211805db323071 : FAILURE in 19m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/29dd25ea01bf47d698d6386de1ea3dad : SUCCESS in 48m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42e49974ef564c6db5284f0b35ad7625 : SUCCESS in 50m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f353b20d0d384507b321d9bfceb6454d : SUCCESS in 45m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6d1aac1d88d04dae955c2aa97dda0c09 : SUCCESS in 34m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8f592107edfd426bbfb8c636ad11ae9e : SUCCESS in 37m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/88ef4482a9a2448c8c9b452e05704dae : SUCCESS in 54m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d5bbf72dd6ae4a72ad4d9ea0f0b43fed : SUCCESS in 53m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3a5e738694da4d2cadf8ffbd5b449ca4 : TIMED_OUT in 2h 02m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b7a52891a351ae1c35b81450f876ee0df1cd8a3","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-30 12:28:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de2524762d1c69ffd903b46890cc9fa16e3e80a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:37:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b99cc8887c4049a9871dadd4364ba1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db2dd9e3c124738ae0da35c15dbcf3f : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53ca93b906614296841398d933ee701a : SUCCESS in 13m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6855ac48ee0646a4aec109eb682c5999 : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c2dce4fcef4e4c9b34a2236c245b29 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/351a68c274ed4c41b56ffa5983d258b9 : FAILURE in 16m 34s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a3ceb33a6328404bb776245f23e4ba9e : SUCCESS in 53m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d4cba47f91a460ab9e07dd14ca8ff61 : SUCCESS in 50m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecd687f632914f328d3f594961d0f8d8 : SUCCESS in 52m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dff847b7e36e4576b2c3bb22ab11d3e7 : SUCCESS in 54m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ac1a5462d5ea4f8c8d6dc41ab477e59a : SUCCESS in 46m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b24d09bb908d407292928ec8c06e025d : SUCCESS in 53m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c0541a28eac84ca8b8817b831dfb8e17 : SUCCESS in 54m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/28e3c15f5a9a45c78cde4ae5d3e64421 : TIMED_OUT in 2h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59d400b818faa2c57f3a83ef79ce6146a2dd468d","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-11-30 15:08:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07c54531c0491b64d9e3a16219ebcc7d1d471360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 17:13:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59f2cd67c76c4c5ea1ed1d2ab3620f2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b91f898eab40b29e58b7a0a9708b68 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4cb5833b0ef4b74ae3d3abc1206ad44 : SUCCESS in 13m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e25c561c8d8045399d4ba0395b69e0aa : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4294cbcd55f4dd1a2feb139c406c8a9 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/968898375b5549a2893041832d6859d0 : SUCCESS in 16m 28s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/57d10fa5237e4b2a9c14d661912919a1 : SUCCESS in 49m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e893dda11d943a494e37a84916d3318 : SUCCESS in 49m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cad80d8c9e3b449c88f4f3786e4dd397 : SUCCESS in 46m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/657aa92a062446cf85294e94ecd6f674 : SUCCESS in 35m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9d73d3f04e7144f5bacbb04409dfa89b : SUCCESS in 35m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3c88b05f27aa469182d83cf4004faf7f : SUCCESS in 55m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f2a24d81a5ad4280ba8a814e92360a01 : SUCCESS in 52m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3fbf7765c89c4bd9a84104f0da97109a : TIMED_OUT in 2h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8811d36191b500cd37bf88d231ed80cf1f4c1cbe","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-12-01 07:53:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01051c83d0ac1cf550378629958b36da65b80b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 10:01:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b89da9ee1cd49ceb2f92a1fd7749460\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b52b54d17794c468ed730f9144d2843 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/421e5568e65f4538b1312e3ee9259a4d : SUCCESS in 11m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3140f4be170480a856425aac8b7110b : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e7a389d9c54008bac8cf7724242bfc : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41ab3c260abe4802b8b9d640383b26c1 : SUCCESS in 16m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/afd9364a8fc04367a1ad6107ee4ab312 : SUCCESS in 44m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac9e1dd172ea4c209ef5679aa4338727 : SUCCESS in 46m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c091e087e3ca4115abcfab499ee6b9c5 : SUCCESS in 50m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/badd99f2c74748c6bc240f7b6623d724 : SUCCESS in 46m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6e8bb0c9d526477d8d9f55fec8328032 : SUCCESS in 47m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/af8651445faf4a27a82577acbbf9c30e : SUCCESS in 54m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a74a5777d390454fad1ae1ef807609a8 : FAILURE in 53m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/aae52efca49846ef8f8b5a5d30f37482 : TIMED_OUT in 2h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4077bfd66fbede739e440bb59dd80834de09e7b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 19:45:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29124\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}],"_revision_number":2},{"id":"093349a27803f790e7f287affafb9d345f16ab7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 20:34:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca09fe7e8814c9c92b20aeb57921ff3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492aecf5581d44e6b0c546a7438bc227 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05963cf9c6b24a9fb11b56ef542c4100 : SUCCESS in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d9b322eafb646fda3b6772a81c9cec4 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1249e3a23c774ce5a22468cf381ea457 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2d38224de3d40689efba6b029083831 : SUCCESS in 14m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a3ae00bfbbf49a0b4325a04b2a814d8 : SUCCESS in 22m 38s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/c6195f9b5798434186a290dcfff1e182 : SUCCESS in 27m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d36e950061449c3b25149107f966c04 : SUCCESS in 29m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f03f734147fa4c55b9135e02a2916261 : SUCCESS in 23m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/439e391951114efe9c92485a7624f7bb : SUCCESS in 41m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/721ed1afac394f60998e8621a2623497 : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":2},{"id":"1ed26fafbceceb38532908a5474807d682cff2a1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 20:38:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"05b19e5081fab3bdf9cf23beeaeb41b703d597d0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-15 22:02:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f26da96e8274cd4bce38a9798a5a389922423df6","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 22:03:38.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_37598\u003e\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":2},{"id":"df79cfb3f3696771e0fd0a7c856fb4e6afe405cb","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 22:03:49.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":2},{"id":"369cccffcea08aa0139e2669c0424566b2fe7164","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-23 06:01:40.000000000","message":"Abandoned\n\nfixed in previous commit","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8566dd4b5e925f20b9a23a78aabc56bccb8cd41b","revisions":{"67ab4963a12f40cfd8840147fd71e8f9ec61fb20":{"kind":"REWORK","_number":1,"created":"2023-11-29 13:12:39.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/41/902141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/41/902141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/902141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/902141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/902141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/41/902141/1"}}},"commit":{"parents":[{"commit":"722b4af6b81f607bc273dcc156d448802697b339","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/722b4af6b81f607bc273dcc156d448802697b339"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-29 13:11:45.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-29 13:11:45.000000000","tz":180},"subject":"Fix broken tests","message":"Fix broken tests\n\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\nChange-Id: I1bb527fe518ecc5548e86ea339fa24e3a76b7e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/67ab4963a12f40cfd8840147fd71e8f9ec61fb20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/67ab4963a12f40cfd8840147fd71e8f9ec61fb20"}]},"branch":"refs/heads/master"},"8566dd4b5e925f20b9a23a78aabc56bccb8cd41b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-15 19:45:07.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/41/902141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/41/902141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/902141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/902141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/41/902141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/41/902141/2"}}},"commit":{"parents":[{"commit":"8d8d71891be4ce4ed24c5516d13020480cc797e6","subject":"Merge \"Allow admin to retrieve all values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d8d71891be4ce4ed24c5516d13020480cc797e6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2023-11-29 13:11:45.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2025-02-15 19:45:07.000000000","tz":0},"subject":"Fix broken tests","message":"Fix broken tests\n\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\nChange-Id: I1bb527fe518ecc5548e86ea339fa24e3a76b7e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8566dd4b5e925f20b9a23a78aabc56bccb8cd41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8566dd4b5e925f20b9a23a78aabc56bccb8cd41b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
