)]}'
{"id":"openstack%2Fmistral~904671","triplet_id":"openstack%2Fmistral~master~I1af648ca92c9becb64e8408558a0c3222905f318","project":"openstack/mistral","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-01-09 18:22:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1af648ca92c9becb64e8408558a0c3222905f318","subject":"Update python classifier in setup.cfg","status":"MERGED","created":"2024-01-04 07:58:05.000000000","updated":"2024-01-09 18:23:49.000000000","submitted":"2024-01-09 18:22:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904671-py3.11","meta_rev_id":"c6d78eb246f1aedc8ea21b7c63993fc31aefbfa4","_number":904671,"virtual_id_number":904671,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-09 18:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2024-01-08 13:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-09 07:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-09 07:55:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 10:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-08 13:35:29.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-01-09 07:55:19.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"8e5958bfbd72daf7b7d021d45bf8d497e354f289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-04 07:58:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ffab07094a4c6c825934f18b5c4a2345f7cf926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 10:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/784a0937a88e4faa81986dcc4959f326\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b58e1c96e9c405cbd4be3226186d581 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb5675d72b9452eba2f84ba51eccf5b : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5c03152e8042c5a4e85f60f95f9775 : SUCCESS in 12m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08392c7833854be0851ac20f88782cc7 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7f2680c883498a8b8627cd55392731 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ba3ed55c9094220ab656a5026d3f7d6 : SUCCESS in 16m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/982f0701dc61427a98a03cf84adda5c2 : SUCCESS in 46m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8e9f8e711804fd182d0e4d7c637bc3d : SUCCESS in 47m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bae92349bb1243b781b02436cbf1b720 : SUCCESS in 53m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/065fa0906b9d45c68719837ba572ac2a : SUCCESS in 52m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f352453e1fa24d9fa140f16e08f5f281 : SUCCESS in 55m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8c2a25da840e4e1fa6bb3ee13aa0fa48 : SUCCESS in 48m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1709260e9bb3418bb57e6db8317da38a : FAILURE in 44m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7ef993101d934ce291ea4f468e05f3d9 : TIMED_OUT in 2h 03m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8120d3a20ec2ce03e4af3e78ef02b7e1a68941a7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-05 04:35:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25226afc6d0f30ab40439887c6e93dddd78a5387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 06:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240daeed301b4ff1bd0cf1d2eee3c87b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6a72b5c42644ebc965d8f830a5e67a2 : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe670d5c2f14d7da5fe4d03aca50ae1 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e32685cd60dd4532ad9b6195071e6843 : SUCCESS in 10m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9488d65a55f04ea78b449aca4ff24632 : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4999d48ac669415aa744fb0156e67e21 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7d9f128ab4647fe99ebba7e97cec1b0 : SUCCESS in 12m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/23588c11b1624cf586d1d25261eb409a : SUCCESS in 27m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09be60f362c41d7be8be950f4be1f24 : SUCCESS in 59m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc69df8f1ed34d92b55495b323961bcd : SUCCESS in 48m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd6cf5ed377441d3bb973d7c9a208d46 : SUCCESS in 29m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1a911fdb0c1f46288f6384877efaa4ac : SUCCESS in 46m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d0b987543569412480acdaeeb4bbc17f : SUCCESS in 53m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3e92660e5ab54d3a82345ae1c00fff5f : SUCCESS in 52m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c505ac959c1741a3976b7181efc31a51 : TIMED_OUT in 2h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3940a44a6e89a5de3f4e2b0c4ee468442949e52f","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-08 13:35:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d720a860481186e0f71e91e4778ce8af108ad78b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-09 07:55:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d65f83c45d9548445fb311f5e0bffe3ed3fa505c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 07:55:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18b2ae9293a66a59e8c32be452775b99f564c611","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 08:58:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1009eea620384ff491411befe53993bf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd082800330849598e275f9b07be5a23 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664df20bd6ec4fd9bcf72efa3900990e : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b19e497f5b344608bc9fc86588ead481 : SUCCESS in 15m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05826378241b4d2a8d1f9cc9003b6cdf : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e948c14fe645481189e287ea44e096e3 : SUCCESS in 7m 48s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5156616c0e549fcac15f02243e560c1 : SUCCESS in 48m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37667cce0d384bcda6717955f469a83e : SUCCESS in 55m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd4f8974b3c34b2790541b6d145f76ca : SUCCESS in 43m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0a5b7ac6d95f40b3bc73a1cb85ca0b52 : SUCCESS in 50m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/60a062a72a474f009916a4bdd2f03117 : SUCCESS in 47m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ceeb334e1be147988b738cada3e09244 : FAILURE in 49m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ef258884c0414345abf3749e524d4a49 : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"73fcd6a4fc0b35af6e921b2cd5eb480eec2e0a2d","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-09 09:41:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86f20635b1dedaa47f5acdeb9d083bc4cc6dd5d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 11:46:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3698a113b384af7b31a759fb7724abc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/44057b24ed3f478eb23372d1f95569af : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef0de57e7494460c83ab183d9cc81eb2 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15273df5079e4858be9b421ebc398c7c : SUCCESS in 12m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a661ebc75e924bfc9b95610463883b04 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd7f039a0464e17a30a7da451291ed6 : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97327a699f3d40268b85059b17e2c2e8 : SUCCESS in 16m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/235411e1002840f99c9fe794edde9c6d : SUCCESS in 1h 03m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46da0c2877574e6796188b0d099eb273 : SUCCESS in 55m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fe59e84142b4b33a8b45e59702dd502 : SUCCESS in 37m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0b56b6a0efca4338926029218300fcfd : SUCCESS in 34m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/18ea618d81dc46c0be85184d63990dd4 : SUCCESS in 33m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/27a49c1088f8415c858bac7716a22e9c : SUCCESS in 47m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/791deaf63f894d31a129efe3ae19e567 : SUCCESS in 44m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7d7a066c15eb4b36b4130e15ac4706a1 : TIMED_OUT in 2h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a12d5d6c7c4fb2f2395d856a302eff4e69583e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 11:46:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17819cbf150c27f50a07ec0e0eece08d4083c2fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 13:11:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1302d06d64a4f06809ac3c5e7e3fa67\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a3ad226dc8744ede8b1614dc2b1de9f2 : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9471c1a20f8417790fac96aa1228d3b : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d80d5e212214c4bb19324e08ef20a9b : SUCCESS in 14m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d04a20712f104d17a50a330069b28d3d : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e6ecd2102f4c599c91c2297685ba14 : SUCCESS in 7m 47s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3a030fcb64d144debbc77826f3b7dd5e : SUCCESS in 1h 03m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/347ee3d9cff946049840893e0b0a87bf : SUCCESS in 1h 01m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/571961ebb87b48dab23c1320087ab96c : SUCCESS in 52m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d4e3d1d3ea5c4eb0a8f1caca0e4c5ada : SUCCESS in 48m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/73f34f99450243d4bc0394a46e0c207f : SUCCESS in 31m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3602930e2e1141968a07234a65fff785 : SUCCESS in 45m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ad9fd07b6d4748bd93287e48ef6127a1 : FAILURE in 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"98f49f16701ff20b6d96dcc7d9d1444a7edf7fac","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-09 15:21:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af099e6341f6417f7fb8e0a7903e50c0d8649ec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 17:24:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4982ba0cd8c34f69a44c5460c2de3ace\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6b42970902e8440398e79191ec422630 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f196a1ec950b42f6896af7ccc08af321 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124b062e85394a0387b62d36df19af5f : SUCCESS in 14m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/309c925a5c3f45c6956248f2f8f331cf : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782a51258681487b994ef409a22f9745 : SUCCESS in 9m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aa2bab1d0be4a05861d857f739576b7 : SUCCESS in 16m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/81ffdf84ff6f48918f01870855165716 : SUCCESS in 51m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4bfa160bc474dfdbff37523db02098b : SUCCESS in 45m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/269fe191453c46ac9a6a845e54a4378c : SUCCESS in 47m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/df9bec9708d94628bb68fba227dc497d : SUCCESS in 47m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/793263106df24b51ac211923fe41226c : SUCCESS in 47m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3af7ffd639b44ecd90b0aa434715fe46 : SUCCESS in 55m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8ba4caaa99e84630b008d31e4d066b48 : SUCCESS in 43m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c703ac338bbc4339a1bd916365d6d8ad : TIMED_OUT in 2h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1faf098ddc59c7f02ef45ed50dcadc9b4cf67ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 17:25:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eed9f14ce7575e11eed4f430cbf5238b4a42e4fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 18:22:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/037f402dcddc4194b67ce44acdec6583\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c78ec61303d4763bff81a2755d42772 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c189d628634ef496d80ba38e6780d0 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c34f8b7f2ba4bc88abbf5fc29fbf7e1 : SUCCESS in 15m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4f2a19ac74d45318ca9b8f14ea6fa77 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b38adaf8544970a59c47535296ded8 : SUCCESS in 7m 25s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73e48e45bcda4067b3211121c889b61a : SUCCESS in 46m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd04fa4c87cd455a9b9e5f90b1e6d97c : SUCCESS in 34m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d91e359d0f1e47d99af04c577acda5b2 : SUCCESS in 51m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f267c3319c6e449d9a9c296c7b2c72cd : SUCCESS in 40m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/701c5bbf7815489e8f129023537f1ea0 : SUCCESS in 38m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5a5ba288afcb41deb519c410e8e04c94 : SUCCESS in 48m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9b9f265760a6475c8e500ae0770df6ff : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"78c684d255c765d2fa0002882756e1ac43b03424","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 18:22:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6d78eb246f1aedc8ea21b7c63993fc31aefbfa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 18:23:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81fad2a96b84d12acd3ebac8abf41b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77cdef33fd14b598f1095ccd115573f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63","revisions":{"fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63":{"kind":"REWORK","_number":1,"created":"2024-01-04 07:58:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/904671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/71/904671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/71/904671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/71/904671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/71/904671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/71/904671/1"}}},"commit":{"parents":[{"commit":"494d210f5836b6611aa5e60f3d28fd79dea83dd8","subject":"Merge \"Deprecate unused [coordination] heartbeat_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/494d210f5836b6611aa5e60f3d28fd79dea83dd8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:15:19.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:58:03.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nChange-Id: I1af648ca92c9becb64e8408558a0c3222905f318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
