)]}'
{"id":"openstack%2Fmistral~905206","triplet_id":"openstack%2Fmistral~master~I024d1d82e32cd2c4d775caaaa8ddd46df7570c94","project":"openstack/mistral","branch":"master","topic":"bug/2048851","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-02-19 13:46:38.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"35916":{"account":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"last_update":"2026-03-28 19:40:07.000000000","reason":"Change was abandoned"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-03-28 19:40:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I024d1d82e32cd2c4d775caaaa8ddd46df7570c94","subject":"Fix: keystoneauth1.exceptions.auth.AuthorizationFailure","status":"ABANDONED","created":"2024-01-10 10:43:30.000000000","updated":"2026-04-14 14:11:38.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"179436a1345ae1ed7cabef90ce865ab5d7e2410c","_number":905206,"virtual_id_number":905206,"owner":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-19 14:38:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":-1,"date":"2025-02-19 13:45:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":39017,"name":"David Abire","display_name":"David A","email":"abire.david@gmail.com","username":"dabire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 13:05:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 01:34:24.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2024-10-07 21:30:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-02-18 13:16:16.000000000","updated_by":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"reviewer":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"state":"CC"},{"updated":"2025-02-19 13:45:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-14 14:11:38.000000000","updated_by":{"_account_id":39017,"name":"David Abire","display_name":"David A","email":"abire.david@gmail.com","username":"dabire"},"reviewer":{"_account_id":39017,"name":"David Abire","display_name":"David A","email":"abire.david@gmail.com","username":"dabire"},"state":"CC"}],"messages":[{"id":"9445cecd83042a06b13d89e2ecefa3677c558e39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"date":"2024-01-10 10:43:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6517e19f3c8c9b2771537f495788cf1e31d8e43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 13:05:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/257a75b407554f8baa6e27a5d20f5079\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975705db942e4386ba90ad405349c47f : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc65904ea9b44badbd440c067bb4ebbe : SUCCESS in 14m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7962b76267074e9b93a944193a4ed8cc : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d59ae694c94e40a8576f51bb455ec0 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d175bfef8d1421489615b00e430907b : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51e14634984d4414845dc0f981d2ebc1 : SUCCESS in 13m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/da903794fd594eac8724141f46515e10 : FAILURE in 35m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f75d4a40d154b36980d0adecb1e5883 : FAILURE in 30m 14s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b228f26d976c4314b7002ac47270fc67 : FAILURE in 50m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/816f415aafa44b8f837500a0e7923222 : FAILURE in 48m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3c040d224da945ccb7881c7e41065bee : FAILURE in 49m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/265e183d272d4f5e87158ec93eda53ad : SUCCESS in 56m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/97e68166e487417f89b7090e109b54a4 : FAILURE in 46m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb2d4025e8e742b6820684281ab42ac8 : TIMED_OUT in 2h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7d5c8afec90d98f7aff777faab0a01a60bf29a0","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-11 01:34:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bec37fa26840d082749c8d5a9b6d8add7562e355","author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"date":"2024-01-11 01:45:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"776606a1c45913105c27cfd6d8e2b67b7bd48c38","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-07 21:30:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd23e74b2ce52d1aa2615f768b6dfb2220ffe4c1","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2025-02-18 13:16:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dc37dd5d60e2ce6922303a4abc13519cdc307e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-19 13:45:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_35916\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"}],"_revision_number":2},{"id":"0eb6509213680bb3f7d6300c666e960743c11eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:38:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dc4f380da1144758ffa4c117c833e22\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2f2b244c6c4851ac9e218e93c74d68 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6ecd2d6b7144f48b7f6f520802ca724 : SUCCESS in 13m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1b41a9873e84f9ab92adb1195520cc1 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c7eff2c682479eb595a4205da08968 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22a97fc6935e4c7197864007372668a4 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6889048c7ecb408c8d6167e6de39d30d : FAILURE in 16m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8bed0865e2aa4f8691e0a78938713d65 : FAILURE in 29m 04s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/8b61120a75974f7abcf866642e54e129 : FAILURE in 32m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8bf15445b404b58ad2381162ba2da8a : FAILURE in 45m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1c6f1ea139b49c38bb014e4135d22e0 : FAILURE in 45m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a29edd4f3352466a8f246a3f9c440dde : FAILURE in 39m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6170da5b99a64f3e8eb7f7f7bec04999 : FAILURE in 45m 11s","accounts_in_message":[],"_revision_number":2},{"id":"57b36baabc23907c3ab2c181c57fd452b65bf735","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-19 14:48:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f47d6a61ba29a6f8721a84e0561c5e1539ccafb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:40:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"179436a1345ae1ed7cabef90ce865ab5d7e2410c","author":{"_account_id":39017,"name":"David Abire","display_name":"David A","email":"abire.david@gmail.com","username":"dabire"},"date":"2026-04-14 14:11:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c374eed8aec087dc1a50834b09feaf16c8550b26","revisions":{"ceb1fc623d24d11b6228145ec8de24842fd076eb":{"kind":"REWORK","_number":1,"created":"2024-01-10 10:43:30.000000000","uploader":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"ref":"refs/changes/06/905206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/905206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/905206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/905206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/905206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/905206/1"}}},"commit":{"parents":[{"commit":"fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63"}]}],"author":{"name":"Pham Le Gia Dai","email":"daiplg@fpt.com","date":"2024-01-10 10:42:51.000000000","tz":420},"committer":{"name":"Pham Le Gia Dai","email":"daiplg@fpt.com","date":"2024-01-10 10:42:51.000000000","tz":420},"subject":"Fix: keystoneauth1.exceptions.auth.AuthorizationFailure","message":"Fix: keystoneauth1.exceptions.auth.AuthorizationFailure\n\nWhen creating a new cron trigger/event trigger, mistral\ntries to create a new trust, the keystone client raised\nexception AuthorizationFailure.\nIt may be related to: https://bugs.launchpad.net/keystone/+bug/1959674\nSwitch from `token` to `admin_token` to make the keystone\nclient session use the provided token directly instead\nof fetching a new one.\n\nCloses-bug: #2048851\nChange-Id: I024d1d82e32cd2c4d775caaaa8ddd46df7570c94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ceb1fc623d24d11b6228145ec8de24842fd076eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ceb1fc623d24d11b6228145ec8de24842fd076eb"}]},"branch":"refs/heads/master"},"c374eed8aec087dc1a50834b09feaf16c8550b26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-19 13:45:18.000000000","uploader":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/06/905206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/905206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/905206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/905206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/905206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/905206/2"}}},"commit":{"parents":[{"commit":"6311d3de404bb513edc05c9f48071382553a7858","subject":"Simplify the TimeoutThreadWithException class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6311d3de404bb513edc05c9f48071382553a7858"}]}],"author":{"name":"Pham Le Gia Dai","email":"daiplg@fpt.com","date":"2024-01-10 10:42:51.000000000","tz":420},"committer":{"name":"Pham Le Gia Dai","email":"daiplg@fpt.com","date":"2025-02-19 13:45:18.000000000","tz":0},"subject":"Fix: keystoneauth1.exceptions.auth.AuthorizationFailure","message":"Fix: keystoneauth1.exceptions.auth.AuthorizationFailure\n\nWhen creating a new cron trigger/event trigger, mistral\ntries to create a new trust, the keystone client raised\nexception AuthorizationFailure.\nIt may be related to: https://bugs.launchpad.net/keystone/+bug/1959674\nSwitch from `token` to `admin_token` to make the keystone\nclient session use the provided token directly instead\nof fetching a new one.\n\nCloses-bug: #2048851\nChange-Id: I024d1d82e32cd2c4d775caaaa8ddd46df7570c94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c374eed8aec087dc1a50834b09feaf16c8550b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c374eed8aec087dc1a50834b09feaf16c8550b26"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
