)]}'
{"id":"openstack%2Fmistral~905893","triplet_id":"openstack%2Fmistral~master~Ie5bed1492de2200d7df35e0ae320a081d897cdae","project":"openstack/mistral","branch":"master","topic":"bump-eventlet","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-17 18:22:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5bed1492de2200d7df35e0ae320a081d897cdae","subject":"bump eventlet to latest version that support python 3.12","status":"MERGED","created":"2024-01-17 15:04:21.000000000","updated":"2024-01-17 23:53:17.000000000","submitted":"2024-01-17 18:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905893-bump-eventlet","meta_rev_id":"98a063d5bccd84174d59d3725945f046e5a30b39","_number":905893,"virtual_id_number":905893,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 18:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2024-01-17 15:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 16:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-17 16:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 15:08:24.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-01-17 16:37:40.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-01-17 17:18:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe1489f7261fa38b1e89e2ad737e8298c8650888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 15:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd0bba8a79b9b047210d6629ea9fd3a11682d620","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-17 15:08:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f6f33d1a64c5cbc1e003615bd5dbfacc185e07e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-17 16:37:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"123d21d333a75f211934cd0d49275658cfe10215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 17:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f917c68863dd4d87bcdd92aef72dc973\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f8c25d0b8649968d799ce9fdd0f8a8 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13d4b97ab745480ead69836b60c08ab7 : SUCCESS in 11m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7eecd6189634678a9f50943f2f842d1 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbabedf8d66401ea07f444659c5c007 : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a77a356df01f475681e6a64300b0fc59 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fd379ca0f0d44bab31845648d07e38d : SUCCESS in 14m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6836b734515346748601d440cd6cfd58 : SUCCESS in 42m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60186549c36c48d6b310fa29181108d3 : SUCCESS in 44m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cff0f7d490b468f960debc5db6a96ee : SUCCESS in 35m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3d9ae1842ff64f8e9ba243352c80eb63 : SUCCESS in 41m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b14bffac96c747649f8129996a02b359 : SUCCESS in 42m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f42dbfafeaf04ba28378ab265acb3e9c : SUCCESS in 51m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c5db2769430f401c85a995f182071122 : SUCCESS in 54m 40s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9064333f58d747108d918f27d114ffe9 : TIMED_OUT in 2h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1ac8d8c01a48bc93a63d24094fdd5a129c27ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 17:20:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf7c35de14398f37b81fa8d0f1abeb99974cedb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:22:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1916b77cca0341e18c50f418975cfda9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba889669eaf340188491431597d4267b : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeee186e17b848519d27480e0cb44a67 : SUCCESS in 14m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0eaab4f9a3984a2298e84478461dcf0d : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f955c427664ef2b63814457b27cc67 : SUCCESS in 6m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/beae5121a73444ef84b411071b4b579a : SUCCESS in 2m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/356134f92cf44fe7b0a003e08437ba52 : SUCCESS in 49m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30d28f524d914e0dbcc394b0a7b00112 : SUCCESS in 47m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac02d58a995648db9c61c2867b0a1e5e : SUCCESS in 49m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bb1ca214cea54ebca9bd1707ae909dd8 : SUCCESS in 48m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8e3a1996b0dc488d8d10a51d5348ba75 : SUCCESS in 46m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5c2521bb40bf45a5a5a0e742f61b7688 : SUCCESS in 56m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f2099341fe6d4ac1bca5fcfe307499f5 : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4cc967d23f4f64dde29722810b240467d1fa0844","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:22:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50c4696f589fe22b744d9c1e2f5a3d14c3bfddca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:24:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/711173660e8641a89733c6d652b0be54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c4beb967484afb9d4502fc16237e04 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"98a063d5bccd84174d59d3725945f046e5a30b39","tag":"autogenerated:gerrit:revert","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-17 23:53:17.000000000","message":"Created a revert of this change as I162f4a6ac10145a1f7a11e64d69f5f7fb3a111b1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0628604428d84bdeaaf76049cfd65f6e976a32b","revisions":{"e0628604428d84bdeaaf76049cfd65f6e976a32b":{"kind":"REWORK","_number":1,"created":"2024-01-17 15:04:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/905893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/905893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/905893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/905893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/905893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/905893/1"}}},"commit":{"parents":[{"commit":"fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fd7ea49fe803bb764d3aef8715ce2c66a8cc0f63"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:04:08.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:04:18.000000000","tz":60},"subject":"bump eventlet to latest version that support python 3.12","message":"bump eventlet to latest version that support python 3.12\n\nChange-Id: Ie5bed1492de2200d7df35e0ae320a081d897cdae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0628604428d84bdeaaf76049cfd65f6e976a32b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0628604428d84bdeaaf76049cfd65f6e976a32b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
