)]}'
{"id":"openstack%2Fmistral~905907","triplet_id":"openstack%2Fmistral~master~I162f4a6ac10145a1f7a11e64d69f5f7fb3a111b1","project":"openstack/mistral","branch":"master","topic":"bump-eventlet","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2024-01-18 08:08:49.000000000","reason":"removed on reply"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-01-18 09:43:36.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-18 09:43:36.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-01-18 09:43:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I162f4a6ac10145a1f7a11e64d69f5f7fb3a111b1","subject":"Revert \"bump eventlet to latest version that support python 3.12\"","status":"MERGED","created":"2024-01-17 23:53:17.000000000","updated":"2024-01-18 09:44:37.000000000","submitted":"2024-01-18 09:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":905893,"submission_id":"905907-bump-eventlet","meta_rev_id":"406a44788e76484bf92fbf658b7ea0eab1156b65","_number":905907,"virtual_id_number":905907,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 09:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2024-01-18 08:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":1,"date":"2024-01-18 08:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 23:53:17.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-01-17 23:53:17.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-17 23:53:17.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-01-17 23:53:17.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73aaa4d3e07ee9d54949154ea30c020da9ae9102","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-17 23:53:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75d60da1ee1d5296b3f1d5f84ebfd6590702fa09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 01:58:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a85657e54ab4a969989a619439157d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a9f39770df4c6395b775cc3423cbb7 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06b9e9c95ee246989788cebffdf9dff7 : SUCCESS in 13m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6dc8075322794e3ab60d28c3308293a9 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1ea9866d5445f38b78373fb30ffd20 : SUCCESS in 7m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99339ee104474055b720da3a5f6dfb02 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2704d9ed8935489c87dac8b444b828fa : SUCCESS in 18m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/95b5e76ca4594a31ab80676408a01774 : SUCCESS in 34m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c47526b2f6e1498abec359eb23a95e74 : SUCCESS in 38m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50a0a08f16f6497ca7d6ad798532fc7d : SUCCESS in 59m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d125d9ae8cba41f9b10ea579cc32e930 : SUCCESS in 50m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8385cacdb5974f0d8f7c2daa2c8fbefd : SUCCESS in 47m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/79a26e0208f14f87979b0e127d9d531c : SUCCESS in 45m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5970d88022ba4ab49f01714e3d48d5db : SUCCESS in 44m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/848746138b6c4f69a662c484aca63340 : TIMED_OUT in 2h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6ccf71795dee62ef54f9ed4f4781999af7f0b34","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-18 08:08:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c55a12e5b771a0c8e1f7037522cbd925423f99ed","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-18 08:28:42.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce0dc3f3a33c2b9ada473a5c9b61aab6f3184f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 08:29:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"769f62659c47db5a28a5dae410c524f156527167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:43:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736d0d0cf5de4da385882e261c785e99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ef2d4e98288458ca037d3b5878a8ab5 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88039715d43d4825ac8e2f90799c197e : SUCCESS in 14m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd7a3f284726403fab5e98dfbb3d9c87 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6714baecaac041abb003ae5fbd42da9d : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec304d826a4f416487efeb9f278b7ae2 : SUCCESS in 2m 50s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a2443fa459704dea888df827205c93f6 : SUCCESS in 52m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/309dd8732d2c4964b5b1775e7a0370e0 : SUCCESS in 35m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad396460bd9848a0b1604836ac6e6107 : SUCCESS in 51m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ac906e70b3ef4081a912fc203e3fd7bc : SUCCESS in 48m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4243e498595f453582c9952392b97667 : SUCCESS in 41m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6623f670abda4d1fbc5093bfd75d3221 : SUCCESS in 47m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fc4f095ad60d46038e6d9cdfdabd6e12 : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"34bce981d2db608466fee1c31e184b3364740923","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:43:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"406a44788e76484bf92fbf658b7ea0eab1156b65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:44:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5662990b7e0c459a9f499488ebd6e3a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8c9d9ad7954ee68ef68ef895887be4 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b68c105696d9c5fe331ba3083b05bbf3c3db31ad","revisions":{"b68c105696d9c5fe331ba3083b05bbf3c3db31ad":{"kind":"REWORK","_number":1,"created":"2024-01-17 23:53:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/07/905907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/07/905907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/905907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/905907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/905907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/07/905907/1"}}},"commit":{"parents":[{"commit":"e0628604428d84bdeaaf76049cfd65f6e976a32b","subject":"bump eventlet to latest version that support python 3.12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0628604428d84bdeaaf76049cfd65f6e976a32b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-01-17 23:53:17.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-01-17 23:53:17.000000000","tz":0},"subject":"Revert \"bump eventlet to latest version that support python 3.12\"","message":"Revert \"bump eventlet to latest version that support python 3.12\"\n\nThis reverts commit e0628604428d84bdeaaf76049cfd65f6e976a32b.\n\nReason for revert: This is too aggressive and needs wider community consultation.  The constraints change is being reverted.\n\nPlease revert ASAP.\n\nChange-Id: I162f4a6ac10145a1f7a11e64d69f5f7fb3a111b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b68c105696d9c5fe331ba3083b05bbf3c3db31ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b68c105696d9c5fe331ba3083b05bbf3c3db31ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
