)]}'
{"id":"openstack%2Fmistral~906221","triplet_id":"openstack%2Fmistral~master~I318a98f4f3034238de808508c000169519c6fa5c","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-23 00:51:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I318a98f4f3034238de808508c000169519c6fa5c","subject":"Move zake to test requirements","status":"MERGED","created":"2024-01-21 14:54:42.000000000","updated":"2024-01-23 04:44:03.000000000","submitted":"2024-01-23 04:42:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906221","meta_rev_id":"046d91d2a9f3fd3c067c3759ed15a0e0ebe103e0","_number":906221,"virtual_id_number":906221,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 04:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2024-01-22 09:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-22 16:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-22 16:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-21 17:05:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 09:52:11.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-01-22 16:25:09.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"1cab04f37d2e8d0b4a8d2e904403f17fb57fd681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 14:54:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb11fbe9cbe121c98ff1d1d7d275f8d16c3dd8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 17:05:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6622c562da64d76b66e91c6a2279a1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae218f4c65e044149cc92e6fe07acd0d : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdd8931a90944661a2816dcc0608d8a0 : SUCCESS in 14m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3866ebeba0dd451ca435e3c553d03f90 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1dcaa5723254a8f9d04161d5103ef04 : SUCCESS in 6m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4715c4ea67242609bfb585c3d12eb35 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/432db4c636394a1ea0ca2139200e28ae : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27817e88c2c841bc9a2c6109eab37540 : SUCCESS in 17m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d4f9b98a5c7b4ea9b6606798d3ad6889 : SUCCESS in 46m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bbedd57ff2942a0b7dcff4c24695d75 : SUCCESS in 48m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7cef07ac96484bae4570a706f54f4c : SUCCESS in 48m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5c1e13d884474609b6cf8c1687391775 : SUCCESS in 49m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc55b3e46bf340bda1949596b7ca399e : SUCCESS in 47m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c6b97ac1d60b4ff79c2c41fa967b71bd : SUCCESS in 58m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4f5aac6faefa44909c6d85197af717c1 : SUCCESS in 54m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/23a84dc323ae483097064eb8b039c97f : TIMED_OUT in 2h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f02e53cfa71714cc646e8076300efc068a6ef6f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 01:10:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5a4f6cde226ab00f6f77400fa31605c9e854d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 03:34:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5fa876b99ba4bae9d339b7a1c154c57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2da0d6956443b7808f7c55ffc8239f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124d1a1b075143c392277065e52a3728 : SUCCESS in 14m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18d02e160e0d40ccaaad16253c93cfad : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0067325c65b46a9949205c9e3fe0659 : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/94f0d22eecc7442a96a9824ab9c5eafe : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8210f621968455f82f6bf6e86413695 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd901e3b62084ff98409f5e64415f42f : SUCCESS in 18m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/93f8c276336846ffb22dba9f995201e7 : SUCCESS in 46m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3055dd0f67ed4f3b804e6a5113d2ee13 : SUCCESS in 46m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68e5508103774639bb3c30919e503000 : SUCCESS in 47m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b4003b189ee9428e9c53559d08852e12 : SUCCESS in 53m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4652ba764eea48888e82c9bdab529b35 : SUCCESS in 48m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d3e069f515ec4ee09ac4cbfeb2e6fb46 : TIMED_OUT in 1h 00m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bfc5052509944d51a64afb13fd9a38f3 : SUCCESS in 52m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8d5bddb9c44a4ee5b7bb3b1fce1fab39 : TIMED_OUT in 2h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0222563ef14062415c93720944ea5199ce9b567","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 03:37:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0218a340a7b6757b6dd8e696cff715ac518ee34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 06:05:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0da95b423e4009b510bdf825a309d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c4b0420cc740869aeaf6813a039f41 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e58706dc0d044431b8f4e6bbbbe85dd1 : SUCCESS in 13m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc2eedc5dbb847219b273816bdd75faa : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3795814582cf48e28deb518f004965da : SUCCESS in 8m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c623ff645992458299888ac7614e59f3 : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0999e82c3ae842a19709119f314634c9 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2056056d6a84d2a814a4df6f2aec9c0 : SUCCESS in 15m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c452eba1f0ee45c7a56f491d155dfb4c : SUCCESS in 37m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15caf8b257124f0a89575f19dc8813a4 : SUCCESS in 52m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64cf415161754389864c4ab0d5fe500a : SUCCESS in 44m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/769478f735134a38835bad417751222b : SUCCESS in 32m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc5c5dcd34bc48279bc72f6ffbe5bf1a : SUCCESS in 38m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3f1f20f48551461cbb2b8f5fddea6b51 : SUCCESS in 47m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a8c61dab94ec44159b79beb7520cd4cc : SUCCESS in 50m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a0b1b8c39524e11951d476010028c47 : TIMED_OUT in 2h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"424eefdfe97297844e65e0dbf0b416dfc357a453","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-22 09:52:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"48c2136e7f97128c92ef5dad364c728790210ab6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-22 16:25:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af72f0df9228c216dc5dd62afa4351e1dd4d4943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:25:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6100660fe3485fa3a6f218a88891acf8e1779473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:29:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b6c1859f7df43998e2c9d8f181700cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f676ab23a37243af91035860ca868256 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d82ed05bd049f783dccce92e7df99d : SUCCESS in 12m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7696fd0f0794e968223acd2f4430be2 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2723c4cef8944d50ae36024a44b928e2 : SUCCESS in 6m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d94ca71a726847d389b90e671a75c25e : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d9f81be6c0045b7bd44298c5778ce61 : SUCCESS in 3m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5a3a7a00418b44d9bd1aa529d6fdff60 : SUCCESS in 57m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78ad4442dd404f69af0b1a1c58edb5a8 : SUCCESS in 37m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d7e769d608845928a4357f3445169ee : SUCCESS in 38m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/10b7b488fb984e71a971604909c28425 : SUCCESS in 34m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bd331862284a44988d6847e52a46fec4 : SUCCESS in 40m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ac87a55d26014f028654cdf0beceed7a : FAILURE in 48m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7884440ec7da4bbbac557f91d7cd775a : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":2},{"id":"324896378969b08c40c8dbe4e0010ad502e050a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 00:51:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86acbcdcef84f790116c423ba679dfed29ee8f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 03:24:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b075d8a9d743e58b9b50a94cdbc7ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78c5035169444d19af864d8f7396b42 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e30901a70a094c5a967132f4a2c07689 : SUCCESS in 14m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1ee07e6eb1f4072a4bc5d2536aaabd4 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ae4b273d3f4443a91e711162fa16ab : SUCCESS in 6m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0d4e079c12c42b5beaab78df1e01cbb : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5142e680fba24c49987708a424af0e88 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b530ec7aaafd41fb9417283f05c59b51 : SUCCESS in 16m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/826b11a7203e41a79a5a5e382f70f1d9 : SUCCESS in 53m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5330bb52d1f645c09e6116d870ce8054 : SUCCESS in 47m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/109c9abc87e947638286a075d2e13eb1 : SUCCESS in 48m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cf624625ed8143bca9152cd2253a2e33 : SUCCESS in 53m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c8d8b786c1e545ffb4a72ddd2e078492 : SUCCESS in 47m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cc2c101e438f4c02be6178375c9f8ae3 : SUCCESS in 44m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c6aa7ffed8548e3ae0b789ba97bd54d : SUCCESS in 52m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c466675f14c646f0a67118281d7d8939 : TIMED_OUT in 2h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b8dc465aad0e47ee2a2a209a8090d3e71c05901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 03:25:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf6ec9437538ed8cc2dc1fa6e9a85837d1b5b04f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 04:42:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7f13384679e4ea9b3f51968c9ceaa1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fd035d94864738a565bd44bf1c5dd3 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9140874c2c945538f8d01d1ccfd2f3a : SUCCESS in 15m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc3eabc4f62e4e1085c926888d559c74 : SUCCESS in 14m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9aae62a73645c581276cdf4f4b957c : SUCCESS in 7m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14ef90f8dc6e4a90ba83128ce209985b : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f40c14e66a494afc8a6ecc645285746b : SUCCESS in 4m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b1daa72a18614dc1b17588c866e14e05 : SUCCESS in 55m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/435fb37b00c34cbe891447292b080555 : SUCCESS in 1h 01m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14e740009d4e4d7c8696c9eb18f78902 : SUCCESS in 38m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/de9ee9ee33764d7aa2074a42053cdd88 : SUCCESS in 49m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7aff807567244387aa308d77db187143 : SUCCESS in 58m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fb9b54ad52fa447b85c5e7fdd8ec463f : SUCCESS in 55m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f686c1805f87435580f401ddd2dbb1ed : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5014c27dc9eff17e9d3bceb75e442d864624a299","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 04:42:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"046d91d2a9f3fd3c067c3759ed15a0e0ebe103e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 04:44:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec08d8037aae4475834cc4aae3791498\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c330db0095483488cfdceb9626536c : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/220f15669c5542f9afc6ec72fd8c11e1 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","revisions":{"d29d39f5599a7fbd32d24bb0ba8303975bedbe77":{"kind":"REWORK","_number":1,"created":"2024-01-21 14:54:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/906221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/21/906221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/21/906221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/21/906221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/21/906221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/21/906221/1"}}},"commit":{"parents":[{"commit":"5be3de2fd4c1abccbecd768fd2d74061e9d14c4d","subject":"Revert \"Temporarily disable failing jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5be3de2fd4c1abccbecd768fd2d74061e9d14c4d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:54:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:54:08.000000000","tz":540},"subject":"Move zake to test requirements","message":"Move zake to test requirements\n\nThe zake library is not actually used by service logic, but it\u0027s only\nused in testing.\n\nChange-Id: I318a98f4f3034238de808508c000169519c6fa5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}]},"branch":"refs/heads/master"},"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-22 01:10:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/906221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/21/906221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/21/906221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/21/906221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/21/906221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/21/906221/2"}}},"commit":{"parents":[{"commit":"b68c105696d9c5fe331ba3083b05bbf3c3db31ad","subject":"Revert \"bump eventlet to latest version that support python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b68c105696d9c5fe331ba3083b05bbf3c3db31ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:54:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 01:05:16.000000000","tz":540},"subject":"Move zake to test requirements","message":"Move zake to test requirements\n\nThe zake library is not actually used by service logic, but it\u0027s only\nused in testing.\n\nChange-Id: I318a98f4f3034238de808508c000169519c6fa5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
