)]}'
{"id":"openstack%2Fmistral~906222","triplet_id":"openstack%2Fmistral~master~I0e835858514d82b181c2898ccb87eb8ef72ad10c","project":"openstack/mistral","branch":"master","topic":"drop-zake","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-29 06:09:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0e835858514d82b181c2898ccb87eb8ef72ad10c","subject":"Stop using deprecated zake coordination driver","status":"ABANDONED","created":"2024-01-21 14:58:10.000000000","updated":"2025-04-29 06:09:09.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"18eb3ecb4d9a23bd06bcc1ac73d5aaf5f9f957fc","_number":906222,"virtual_id_number":906222,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-15 16:45:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-21 17:09:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 09:51:36.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"CC"},{"updated":"2024-10-07 18:56:56.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-02-15 16:04:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"6190552eb46da1b0ded080229ca9ab142613b3bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 14:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"547d57f014a6196c95bddbb682c62e663f974ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 14:58:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e96a835aa923087c210b88469fac5938a5812ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 15:00:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1205023f3e839a556647a44717a65e6a37a6eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 17:09:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a7ca2929a254d9e91acc8d110b223b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664c0364b2d740d688ad59e72578e39d : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d53752a6714710ac71658318e2a296 : FAILURE in 13m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30cd58e6f7464a7eae7c3d3ecd260e31 : FAILURE in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ed29b71b974d6093ebde38e070d045 : SUCCESS in 7m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7331b1e6b59744cfaa05b91e0fb019a7 : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a9b1ebc29ad41a0b2420aa7a3ce427c : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3917264858d4ac7ad3a97abf0f9d8c3 : FAILURE in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ab232b9a8ef243c680677c9583f18484 : SUCCESS in 28m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3987c7385d9b426d98fda8f54e31dc97 : SUCCESS in 30m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b57d882c357548859c419efb5253ad3d : SUCCESS in 29m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c170df575e54a988d4f14df6d21598c : SUCCESS in 28m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7f5f56d09a9147108c7ff72159bbdc0f : SUCCESS in 47m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/65a77140defd4fa89568c08c7b0eec1f : FAILURE in 6m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0304f9a3de3d41e8a6facf0f8d824f5e : FAILURE in 5m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bfe66c42c871494790435d57f324b376 : TIMED_OUT in 2h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"418391587d872f93a3fb9173788c90fb95bdd068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 17:25:49.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"275b29a6cc02bb624bb6be913259e995ecbda805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 18:19:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"26b3d15e12486f8b89321b10b955ee7e5db33cd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 20:27:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12facfb133844cc196513d15e0ccf659\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995a828d6a424c53abfd2653354b32ff : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78f995a53fc04a5aab44ce1e4e3102c5 : FAILURE in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d98d202a04544db693c2e7ea292a5d6d : FAILURE in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78587c454924476884521156c5c7d238 : SUCCESS in 6m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d05c663c31384903954a8a3d4edb4c15 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0cc1a8c96642f5b9547dfccefe18c4 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e7ca4e26f60447bb388098760168b34 : FAILURE in 16m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/42dda8b8695f4dafa9a1a1d9aa1974f5 : SUCCESS in 40m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f98187d53e64561a09661a36ad3744a : SUCCESS in 33m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ee32c8e092c41b7b91ee994524d1195 : SUCCESS in 39m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2f3e58f46ec4412282554b07c9650982 : SUCCESS in 39m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5b8755a783f14c159207671b2f0f0ff4 : SUCCESS in 49m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/151bd45e7ca440f4bd80bc756a678813 : FAILURE in 5m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6a3b8486a12f4b05ae976843ce985f80 : FAILURE in 5m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/91ee27338a044cf6821908d2209df04f : TIMED_OUT in 2h 02m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"519b78f542b45fbd23e305dd7a992e11d5eb8d1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 01:10:32.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"69238a1939f4b56ed4a926648fc9e5368968ee08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 03:18:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb73aa0b1cb542b991d096d6b92b54d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132fe992b05148609c95a97e2ed92843 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68cc3418967e4104bcba3b96e42a95ee : FAILURE in 10m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ece0859b89c4aa18272c76292c381a7 : FAILURE in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b7e0697cdc4ae299ffb5d5681748b1 : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc992367693148cabb27cc872076e201 : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9978e52452643888fb50210700a8266 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/631ddb900d6849bba5679d084786c55c : FAILURE in 19m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/31fdc6aba10a47008598c69581613324 : SUCCESS in 31m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ed5d844a75d4d4992d0afd783de1e0a : SUCCESS in 45m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b70b140c244846ec9679ab126697c477 : SUCCESS in 49m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/743cb1708ddd46a4b5ada689812fab0e : SUCCESS in 31m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2200ff28547c426a88fcf6f4e2c44a83 : SUCCESS in 30m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/77792c3ff8894d1596d37fe5e3cc08b3 : FAILURE in 8m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/02f0543930da47738d96e8a25377cc1a : FAILURE in 5m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9fa7d34aaa6b4dd18373e2d899626d3e : TIMED_OUT in 2h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"69991916a769ed9bf9ba036deca4c640a2fa142e","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-22 09:51:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"98b6853c5002bb5f57b51922dfbdc94ad5376b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 12:00:12.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39d124eca4cb4964a7385c6281cbcab1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277fc6a593244cd2bf24c38f6fe73dd5 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c875c1936e44017933fda4f852c0666 : FAILURE in 16m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cf7467c8d6a4be5b4067b3d097febf1 : FAILURE in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b037cd7855409eb47bb3027b3ed665 : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d72305c2161e43a4aa714f767c90b606 : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435625f38b594696a684df04b169f38c : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eacdb5bb7069463a805aec0a9d00f474 : FAILURE in 17m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/028fc95e365b4090ae18fc644ac202ab : SUCCESS in 32m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/342e36c5dddd4816ac2fe0ea430bd19b : SUCCESS in 46m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83b33c79fd004d0da90d81e9d836afef : SUCCESS in 48m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd222d0e5f944abea471b358cd372af0 : SUCCESS in 50m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/59ddeba5861547a8bb970c85120a9aaa : SUCCESS in 50m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bb05c50ce61b487ea1ff7fe78f0066ab : FAILURE in 5m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/67240121830d4d19b14668d6787579c9 : FAILURE in 5m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a6a014bb22894e58a640a5cb39a93181 : TIMED_OUT in 2h 02m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"348da24d82367483d1e5cc75f2d38772a170b659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 01:16:00.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"17a6ab174010caffbf7106cd48bf2fa5f0727838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 03:21:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fff246dd74f74845abaafde67be76564\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f60e503f8814fbfa4ac651ec575c8b5 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bcd5d1deee0453e8c1871596d93aa5c : FAILURE in 12m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57b3056c6962435db18b603e9fd09da2 : FAILURE in 12m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8348f9d137b84f7fa63dc54b613f4681 : FAILURE in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3019f3bcbb48c589e551f39c6ffc7a : SUCCESS in 5m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/235ecf0661dc4fa7ace4d440a9a461e6 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/500178745bde43ff95e8b36175eb9908 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/932a3983cba646b2830b84a91dec151c : FAILURE in 12m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6404190dbb964491958dd11f9d470823 : SUCCESS in 31m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/600df1d1a7024dd0bd7f913c10fc53e5 : SUCCESS in 46m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53d15ff8a5dd419693b9e9bed5a1f42e : SUCCESS in 30m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bff314b63af24bd0b1d7e4f5583d9315 : SUCCESS in 28m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/647b8c8ba4f6456ca58864e36b595c06 : SUCCESS in 28m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/46ade36bbff74807821a8dd1ca910226 : FAILURE in 5m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4faafb4c88bb4acfacde64a405c04d4b : FAILURE in 5m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f9ff19eb328540cab0a5c82f5074ff3a : TIMED_OUT in 2h 03m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f147ead55c0d738099e7e3ce526132d810426baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 16:23:49.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4f6b7fc28d90d5ea2ad0ba6ed0d12010bb2dbefd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 19:15:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/319daf12cbee4aec95ffa7ee7389d9f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77607b9650d1413381a3af6a7ba064fa : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/192953edc31143e48fbb7c449266dbc1 : FAILURE in 11m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c35bc3042c82445fa11866ee915c3f61 : FAILURE in 11m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2a6088bb6bd4a3caf5605d1f38542bc : FAILURE in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb5127d371246fd83548cba6e8f9a86 : SUCCESS in 6m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/304976767d7445bc97b2aeed41eca307 : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ed529d902a4dd483b189f6ece164bc : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ee3606b816046c98c401e7c11851db8 : FAILURE in 13m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cdab3c9c9b594f19b84985ebbfea3ac2 : SUCCESS in 40m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f76f066097f4194ae5ef6333d9d888c : SUCCESS in 42m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c0957012d1b4b299dcff26f2dc7c1e3 : SUCCESS in 37m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6d34a2ff419a42da837486ada49c0368 : SUCCESS in 47m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1281491ce7c242a7a149b9fc38688312 : SUCCESS in 47m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8169745d5b744327ae2f9384d6312416 : FAILURE in 6m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/75ba20d042ae47f5997f84a159aa07c2 : FAILURE in 6m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1217668e6aea4b218ca331a1213d0562 : TIMED_OUT in 2h 03m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"de6247d96d77711ea08cafc84501b27bb356be18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-07 18:50:16.000000000","message":"Patch Set 9: Patch Set 8 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":9},{"id":"3eed589439ba34ca5bcd87d0c6746a8e9bf5b9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 19:39:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fe2882929ae4f58bec4d054d9d9f3bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41eef29bb1534914a811a62fb0aca7df : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/108805c1e21f4715911531e5d79b6829 : FAILURE in 12m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcd1e51e74f74248ad9b1f1d7613db79 : FAILURE in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f958886b1ad4acb976fc805b1f89761 : SUCCESS in 6m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/df7840f75b714290a3b1d2da477ea4bf : SUCCESS in 1m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/284b5f893250445ea2c9fbb8a9e592d7 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c292730514a498b8ace7402ab496eab : FAILURE in 14m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/27b2f98871c049e3850c40cdbb683063 : SUCCESS in 25m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82cdc33f132c45ffb5987d8ed2445f10 : SUCCESS in 23m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c81baa2acab4c30ac6361a4114f8d1b : SUCCESS in 42m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6a36539757d049f7a5f7ce0941b27d43 : SUCCESS in 47m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/68a758e34e3744bdbe400003b4cfa008 : SUCCESS in 42m 46s","accounts_in_message":[],"_revision_number":9},{"id":"f5a049eca2315d6c1fe394bb0a1274ce7b93a183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-11-06 23:55:51.000000000","message":"Patch Set 10: Patch Set 9 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":10},{"id":"39d733a825c450ac9d92818b0d5bcbbe17629c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:31:33.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afad395df4e147c287152fd67d56f613\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873e04d349344c6b9d9076970b612b6d : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90b1cdf0f94b4a6bb63e8729c23008bd : FAILURE in 12m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c72cfdab3ef4f4ebf7c4826ad868c9f : FAILURE in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56478f1cb1d45fd9a1173529118e092 : SUCCESS in 6m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac741918a3ae43a9b806819e5ece20da : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba05b64f1e2045c18e1febb0a6933c39 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0f54c2ef9946c6b10bc53ca1754c89 : FAILURE in 13m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/46bd3a95c1fb4cda9b56f636f726c3fd : SUCCESS in 31m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35436c70dd75410da8835adb2708b5fd : SUCCESS in 33m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c556b1c233848adb211bc10c8feaea9 : SUCCESS in 29m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/031ccc6bf7504833b1dd0b816bf8b883 : SUCCESS in 21m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/13ede1d2249c49eb8f1cf1bdf21decf9 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":10},{"id":"3fcdcefa8205dc07240629617524333ba05f0aab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 16:03:58.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":11},{"id":"799c32884de1b311dc831f2e55fe9368e469f1e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:45:08.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cecc07fa835146ab8f93dd756f16a233\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d22f49590d47fd8096275f386abbc6 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5595987140264b01bb3e8b8c417496b6 : FAILURE in 9m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37fbae2c90bb489da83f76da3ac2fa0c : FAILURE in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476e0ef087dd4285ba0e3794c5b70ee5 : SUCCESS in 5m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/798d104b86ab45ef858facec52b82d2e : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/129e37287fb541a1a70ad68c2c6afc69 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93f97eeea559483a9f5fdd44a0b357eb : FAILURE in 12m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6d54579c3bda47b6b08ef73a68abe738 : SUCCESS in 31m 18s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/193505a570fb495cb88188e1675ac260 : SUCCESS in 27m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bedd8a76c41447392dada059afd95b9 : SUCCESS in 28m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b31a4273c4e04708a6c6a2aa8dd1ba1e : SUCCESS in 36m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a96f6da95c2946a9bc40ddb792b1e7af : SUCCESS in 34m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/202453a57c50419cba8eeb2439f38207 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":11},{"id":"b1254395d1ab1df112ebb9410d7e71e8ba73b897","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-25 19:48:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"18eb3ecb4d9a23bd06bcc1ac73d5aaf5f9f957fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-29 06:09:09.000000000","message":"Abandoned\n\ndone in another change","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"95807386ec5036d3ef6e5fd9d12b83cb2223686a","revisions":{"ad568a7938c443777dcfa9fa9bd617f3e218f37b":{"kind":"REWORK","_number":1,"created":"2024-01-21 14:58:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/1"}}},"commit":{"parents":[{"commit":"d29d39f5599a7fbd32d24bb0ba8303975bedbe77","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad568a7938c443777dcfa9fa9bd617f3e218f37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad568a7938c443777dcfa9fa9bd617f3e218f37b"}]},"branch":"refs/heads/master"},"14aaed1b9edc90fabd403be71a8e16a4e3228750":{"kind":"REWORK","_number":2,"created":"2024-01-21 14:58:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/2"}}},"commit":{"parents":[{"commit":"d29d39f5599a7fbd32d24bb0ba8303975bedbe77","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:58:18.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14aaed1b9edc90fabd403be71a8e16a4e3228750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14aaed1b9edc90fabd403be71a8e16a4e3228750"}]},"branch":"refs/heads/master"},"3aabd3581fdabe07b721eb921c6e53072b0dbe02":{"kind":"REWORK","_number":3,"created":"2024-01-21 15:00:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/3"}}},"commit":{"parents":[{"commit":"d29d39f5599a7fbd32d24bb0ba8303975bedbe77","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 15:00:48.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3aabd3581fdabe07b721eb921c6e53072b0dbe02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3aabd3581fdabe07b721eb921c6e53072b0dbe02"}]},"branch":"refs/heads/master"},"f97276ed4a892dfb902ea7da4639acd012441a7c":{"kind":"NO_CHANGE","_number":4,"created":"2024-01-21 17:25:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/4"}}},"commit":{"parents":[{"commit":"d29d39f5599a7fbd32d24bb0ba8303975bedbe77","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 17:24:19.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f97276ed4a892dfb902ea7da4639acd012441a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f97276ed4a892dfb902ea7da4639acd012441a7c"}]},"branch":"refs/heads/master"},"0f71235749296e50bb50be9dda1bbcb82b7a00ed":{"kind":"REWORK","_number":5,"created":"2024-01-21 18:19:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/5"}}},"commit":{"parents":[{"commit":"d29d39f5599a7fbd32d24bb0ba8303975bedbe77","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d29d39f5599a7fbd32d24bb0ba8303975bedbe77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 18:19:24.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f71235749296e50bb50be9dda1bbcb82b7a00ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f71235749296e50bb50be9dda1bbcb82b7a00ed"}]},"branch":"refs/heads/master"},"01810f1c1673c532d1ae8259d3f651f8951b0278":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-01-22 01:10:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/6"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 01:05:16.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/01810f1c1673c532d1ae8259d3f651f8951b0278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/01810f1c1673c532d1ae8259d3f651f8951b0278"}]},"branch":"refs/heads/master"},"327b53ae2437134c0e299e34fcafd9d61892aeb2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-04-18 01:16:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/7"}}},"commit":{"parents":[{"commit":"cadfe9dca2490afc05d2c2489db700a417cd0f0d","subject":"Fix maintence mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cadfe9dca2490afc05d2c2489db700a417cd0f0d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-18 01:16:00.000000000","tz":0},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/327b53ae2437134c0e299e34fcafd9d61892aeb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/327b53ae2437134c0e299e34fcafd9d61892aeb2"}]},"branch":"refs/heads/master","description":"Rebase"},"d1cc19d006ae3a1c16db8238ba8f4158dc2b39de":{"kind":"REWORK","_number":8,"created":"2024-05-02 16:23:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/906222/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/8"}}},"commit":{"parents":[{"commit":"cadfe9dca2490afc05d2c2489db700a417cd0f0d","subject":"Fix maintence mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cadfe9dca2490afc05d2c2489db700a417cd0f0d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 16:22:29.000000000","tz":540},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1cc19d006ae3a1c16db8238ba8f4158dc2b39de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d1cc19d006ae3a1c16db8238ba8f4158dc2b39de"}]},"branch":"refs/heads/master"},"129da90f0d1bd1fb5f59931d24295d8110838119":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-07 18:50:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/22/906222/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/9"}}},"commit":{"parents":[{"commit":"e71ae064dc3f15c300b30f80f9e700a33282a8ec","subject":"Remove remaining reference to oslosphinx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e71ae064dc3f15c300b30f80f9e700a33282a8ec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 18:50:16.000000000","tz":0},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/129da90f0d1bd1fb5f59931d24295d8110838119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/129da90f0d1bd1fb5f59931d24295d8110838119"}]},"branch":"refs/heads/master","description":"Rebase"},"520315b155c355960c2e6281b424910b0f432cf5":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-11-06 23:55:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/22/906222/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/10"}}},"commit":{"parents":[{"commit":"570cb8928508cb369b955d6c3acc57e8025c8005","subject":"[doc] fix working dir to run tempest tests for mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/570cb8928508cb369b955d6c3acc57e8025c8005"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-06 23:55:51.000000000","tz":0},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/520315b155c355960c2e6281b424910b0f432cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/520315b155c355960c2e6281b424910b0f432cf5"}]},"branch":"refs/heads/master","description":"Rebase"},"95807386ec5036d3ef6e5fd9d12b83cb2223686a":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-02-15 16:03:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/906222/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/906222/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/906222/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/906222/11"}}},"commit":{"parents":[{"commit":"c14c59248b3f873f0b75a868d158f57db792bc5d","subject":"Stop using eventlet to start services in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c14c59248b3f873f0b75a868d158f57db792bc5d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:57:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-15 16:03:58.000000000","tz":0},"subject":"Stop using deprecated zake coordination driver","message":"Stop using deprecated zake coordination driver\n\nThe zake driver in tooz was deprecated[1] because the zake library has\nbeen archived[2].\n\n[1] 9acbc5391521155b04aa6654fb91c90b467b5750\n[2] https://github.com/yahoo/Zake\n\nChange-Id: I0e835858514d82b181c2898ccb87eb8ef72ad10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95807386ec5036d3ef6e5fd9d12b83cb2223686a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95807386ec5036d3ef6e5fd9d12b83cb2223686a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
