)]}'
{"id":"openstack%2Fmistral~906437","triplet_id":"openstack%2Fmistral~master~I05d850344c653bc858b4565673606531da803342","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2024-02-16 15:07:57.000000000","reason":"Change was abandoned"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-02-16 15:07:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I05d850344c653bc858b4565673606531da803342","subject":"Fix some race conditions in mistral CI","status":"ABANDONED","created":"2024-01-23 23:39:24.000000000","updated":"2024-02-16 15:07:57.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48b5737023278e66e9dde223c83064866d9e71c2","_number":906437,"virtual_id_number":906437,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-25 00:51:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 01:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 10:24:18.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"4d35eed3730a7490c93d6ec370376058c2830bac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-23 23:39:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4518d5cbdd92947bede32489e61699cd08f6c01f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 01:44:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47d8d249293e49698515150903ce0e67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fbb2b6fb4f141739958f47720837da8 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c845231f105f4a36bccd36ff453fe447 : SUCCESS in 12m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cd7d3d0e7654e7caa2cceea9ec7d3a3 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee21358f104a4a8e92fa0eb94c5a32b9 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64db655189044589a7f50d0b917c2ad3 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e2106aa32744b65b811f47d32914038 : SUCCESS in 16m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0f618467a94b4edf9ffbe132c90436cf : SUCCESS in 31m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15bfc7dd86624fd984cd4d02620f0391 : SUCCESS in 49m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92d043a7398d46a4bf9e34933fd47af3 : SUCCESS in 50m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0d189d63820b4282b4f1a226e4d20bba : SUCCESS in 49m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/91197b1cbb184d52a5b35e457ab56873 : SUCCESS in 50m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b75e9600aacd49c7b65649521206bd37 : SUCCESS in 53m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a5ecac9f60184576bdc99519bfc7578a : FAILURE in 53m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/623bafa4501940038aac00ccd9a47b33 : TIMED_OUT in 2h 03m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2dd461b5967eb6eca346105965cee2c724f1edb4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-24 07:20:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93668a32edc0bfc0a3dbef8882d776cdbe092be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-24 07:30:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5184fe593958020ca9ae8671aa11169654141e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 09:36:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee8ac0b802144cfa6797d2610620550\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef906590a25448496f071dfa6e32e11 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0c82e8f62bb416bab49393e0f8b1d19 : SUCCESS in 14m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2233e2354ede4f11a58fa4b54fb60cd0 : SUCCESS in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85969a6841684a489ec37bb28ef091a1 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4bf05275f39448190d6d018cc10112d : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b715155e2e094e45941eab72bd4c9e21 : SUCCESS in 17m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/adb4006fe4c749b9b37a55ff553c8ed1 : SUCCESS in 49m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c0687016de24320869b9d5adcd7e155 : SUCCESS in 28m 44s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28e2873bfb304cad86cedc7637f80298 : SUCCESS in 45m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/304f614a6cac47fdb3d518bc268d4ebe : SUCCESS in 46m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/362bda06aad7428aa9baf741c8dd57ba : SUCCESS in 49m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ef048f5159704b81bf2311b3d085f6d5 : SUCCESS in 46m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9070ce302027412cbacd375143b842bc : SUCCESS in 41m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2b1333b4ff914b1d94486ae4793cfff6 : TIMED_OUT in 2h 04m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64cdfa2a90e3ec17f9936ecff0836ac2da7acedc","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-24 10:24:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d233add4150f55fd006ee4bddd479764dede87f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-24 17:53:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0f0242669e17e5edaf4346d983d0f44253717505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 17:53:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"505f92129788467f0fa5ce1cdde8ba5bfb76fb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 18:54:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb480b7fef9c4385a887961326d20f54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed8f340701c45f8a76c2b05a425a71d : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a5021df22c4513892c9f1bdb8f006c : SUCCESS in 20m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9d63d48bf2148a88b218f8a2d9d26d6 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21796e9f3ffb429791f059382fda70b4 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c85b23f344d4aac81a3b8bb51e7f553 : SUCCESS in 3m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c74a2e46f9fc490ea4bebaefb858d499 : SUCCESS in 51m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f77f8c9536c45f39d6004498de3becb : SUCCESS in 54m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5968f7ca0ab9496183880326f7462854 : SUCCESS in 47m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e2134511e4284c2f8404c3128e182ca6 : SUCCESS in 47m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2b424892f9f048b98b8eaa6a07b1da04 : SUCCESS in 53m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/eec6824c83b848d095dc6b7d9971e774 : FAILURE in 54m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/568d738c3141490dbe0c25a26b13f9f2 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c1ef6627a5dbab12782a7b6d588c2ada114cd091","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-24 22:44:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"56a7233424ebf9186e319a677dfea77d115a920e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-24 22:45:23.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5800b44708c3e62b70d20a95cebba4f1d3a5ab11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 00:51:21.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15e728524a2a43a793da000cbbf2e143\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431cda011c5b4bd093dffd629110a400 : FAILURE in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60809e181ba5458ca16ecb26fe3e1404 : SUCCESS in 13m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cb81e2514e349d5ad463b887da89751 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2864439b704598b80866def2619a27 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61cc18a034a64254806157a8baa365d1 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2d1d5a8f889426ba89e4d6e428f9ec7 : SUCCESS in 16m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/01b0704567594d0daae12e2dd3100df3 : SUCCESS in 42m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da13b3a9f2354fc4a0454b64b45742ac : SUCCESS in 31m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fae208ccced24991873829dd03f2a277 : SUCCESS in 51m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3698c6fdb0ce487da314822805468519 : SUCCESS in 37m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/44e8aed136b04ff08a6cbba811700ab2 : SUCCESS in 42m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/28f4e4623ec64de7836e16582a67a1ad : FAILURE in 43m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/449d026549a04ba08704edcf4ab65bbf : FAILURE in 45m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fbbd3c79668e41868c085d7a89a714d6 : TIMED_OUT in 2h 03m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"48b5737023278e66e9dde223c83064866d9e71c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 15:07:57.000000000","message":"Abandoned\n\ndup https://review.opendev.org/c/openstack/mistral/+/906595","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5b088a7ad6671e13324697d5201586747cb09f48","revisions":{"9b7218e358a33538845b19504c4102899379f90d":{"kind":"REWORK","_number":1,"created":"2024-01-23 23:39:24.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/37/906437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/37/906437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/37/906437/1"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"subject":"Fix 2 races conditions in mistral CI","message":"Fix 2 races conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b7218e358a33538845b19504c4102899379f90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9b7218e358a33538845b19504c4102899379f90d"}]},"branch":"refs/heads/master"},"2a1c05bb63bce980a4733b46d064488f7f16c2ef":{"kind":"REWORK","_number":2,"created":"2024-01-24 07:30:14.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/37/906437/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/37/906437/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/37/906437/2"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-24 07:30:05.000000000","tz":60},"subject":"Fix 2 races conditions in mistral CI","message":"Fix 2 races conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a1c05bb63bce980a4733b46d064488f7f16c2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2a1c05bb63bce980a4733b46d064488f7f16c2ef"}]},"branch":"refs/heads/master"},"1bd4a8f0fe2ac5d74905b493f5086457ae614783":{"kind":"REWORK","_number":3,"created":"2024-01-24 22:44:58.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/37/906437/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/37/906437/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/37/906437/3"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-24 22:44:47.000000000","tz":60},"subject":"Fix 2 races conditions in mistral CI","message":"Fix 2 races conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bd4a8f0fe2ac5d74905b493f5086457ae614783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bd4a8f0fe2ac5d74905b493f5086457ae614783"}]},"branch":"refs/heads/master"},"5b088a7ad6671e13324697d5201586747cb09f48":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-01-24 22:45:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/37/906437/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/37/906437/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/37/906437/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/37/906437/4"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-24 22:45:11.000000000","tz":60},"subject":"Fix some race conditions in mistral CI","message":"Fix some race conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b088a7ad6671e13324697d5201586747cb09f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b088a7ad6671e13324697d5201586747cb09f48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
