)]}'
{"id":"openstack%2Fmistral~906473","triplet_id":"openstack%2Fmistral~master~Ia85ea13923216d090ef1e4deee7840e70f9185b4","project":"openstack/mistral","branch":"master","topic":"python-3.12","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2024-03-19 12:53:43.000000000","reason":"\u003cGERRIT_ACCOUNT_35600\u003e replied on the change","reason_account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-03-20 09:29:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia85ea13923216d090ef1e4deee7840e70f9185b4","subject":"Python 3.12: cast random delays as int","status":"MERGED","created":"2024-01-24 09:11:12.000000000","updated":"2024-03-20 09:30:37.000000000","submitted":"2024-03-20 09:29:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906473-python-3.12","meta_rev_id":"7b21f0a6e835500648dd42b90cf0e040cac088d3","_number":906473,"virtual_id_number":906473,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-20 09:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2024-03-19 12:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2024-03-20 08:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2024-03-20 08:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 10:25:08.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-01-24 11:19:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-20 08:23:24.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"6dad1187de02e40d3445582878e90567020ccdc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-01-24 09:11:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4f887d898ceda0ae6b6c1c064dcc8ce1f349281","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-24 10:25:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7621435b4d6f6e7b915ca42158e05fdbbaf63c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:19:39.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/818b619a02a144dda2b02eacb40e921e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d713d498162e46c5a1dc6f76b9c8134d : FAILURE in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/786c9795d7704900b5c231841dc5bec7 : SUCCESS in 12m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05d83f8212db47fa9baaf9e71b0877cc : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e799c2e5d4824f4988a294df4e8057cc : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3def14972f04cf8b28362ef24d97450 : SUCCESS in 17m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/825f7dafb3d946bd925874005fbed809 : SUCCESS in 34m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6baff14b71ca429c87f00b19829bca9a : SUCCESS in 31m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2141ad64745a4587a3ed64917bbd6067 : SUCCESS in 41m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fcd15923c67445e49524f4a0f9c90d0b : SUCCESS in 47m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/523d227c828d4c6bb63da6fe6cc48b26 : SUCCESS in 49m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bd7375caecf849079bd773d7c10d7bf4 : FAILURE in 46m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0ce1694d072742edafb9ad2dfe7523cf : FAILURE in 51m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/44e99f1fe01e4985bfb0e934cc434252 : TIMED_OUT in 2h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54b29dd4c17fb8c153954d25c68d391d6dbf1a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-03-19 08:05:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"272ae4679bb0d3d038a28d65377bae96107eb7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 10:08:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13f3d54e8930407cb53f0b2936060647\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fca80bf902b41dd852f4b4d75231f5a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d0ee084a7704cb38eb68b25483a755e : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e9ff1c991f64048bcff07009fcb15b3 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ed18ca2fbc46a881ec1f092501eabc : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e07e3d4481754218a555885deb04fdd1 : SUCCESS in 11m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ca92b8a0ee324a1cb7d921f082e492b3 : SUCCESS in 46m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9f0b00a744247708427087f4a5e1a7d : SUCCESS in 47m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1eabac51436e409e8ca396c8da4e4bd2 : SUCCESS in 34m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4f181c789ea24e8f8f577663c56f031c : SUCCESS in 46m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/43c15e57c0834f09a9a9c5507fa85778 : SUCCESS in 48m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/34761ddfbef3447c8db6b62c7f2119da : SUCCESS in 58m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/16f40c1de3f743a7bd4e60978192385f : SUCCESS in 38m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1f56db137fc0485b8ad03d537ac698ae : TIMED_OUT in 2h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84fe263beaeaf62ef08e5087c8a9b7695a335b75","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-03-19 12:53:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4d4d12d7d29cc93ce5ea6b75bb14b38211b6902e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2024-03-20 08:23:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e80564c2c81cab7575290fe35b739ae37ab313d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 08:23:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8331de2bc2cbf960d121acc7ff7de50e56ac9c52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 09:29:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f121b69740fb49cba6e8a5457cb24e12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84da9fd66ce4339994a4f871b75eda2 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64ac88e0ea2d4d2e8777cef0c0799ef4 : SUCCESS in 14m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98acbf98b37c4791b08bc244a1f7606d : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4bcd99013a46739c76b2d05f936027 : SUCCESS in 7m 17s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f11c8ed0033f4d8a8197a9879a2199ba : SUCCESS in 46m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3560a761514dca97e638f2adcad9cc : SUCCESS in 45m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/020c70aa025a4b4bbf32d47e3b3dd6c3 : SUCCESS in 46m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4ab0b4f3eff54dbe92e3c95ab18230bd : SUCCESS in 47m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f00022b99b364f81bad65bba6cd006fd : SUCCESS in 45m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/96af6a7803fa4b64a60b28b14593d3bc : SUCCESS in 49m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f45d66f7668349049cea3803dbd54cb9 : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6ed620b3569137978f627674ef6de9969cf44507","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 09:29:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7b21f0a6e835500648dd42b90cf0e040cac088d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 09:30:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492df9178a044bc6807cda257b58dce7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25883a4894eb42c4bcf2c2bb4104e520 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b2c8759d4ef8632ffab1c5647ea3a2ff051e301","revisions":{"beac16281a2fec8ec6267a9ed82a1bf1d075abfb":{"kind":"REWORK","_number":1,"created":"2024-01-24 09:11:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/73/906473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/906473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/906473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/906473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/906473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/906473/1"}}},"commit":{"parents":[{"commit":"08b5a9b7c617e7605f9048769af60ed94bebd825","subject":"Merge \"remove reference to old unittest2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08b5a9b7c617e7605f9048769af60ed94bebd825"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-24 09:09:10.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-24 09:10:45.000000000","tz":60},"subject":"Python 3.12: cast random delays as int","message":"Python 3.12: cast random delays as int\n\nThis patch fixes 128 unit test failures under Python 3.12 when\nbuilding the Debian package under Unstable. Without this patch,\nI\u0027m getting:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python3.12/threading.py\", line 1073, in _bootstrap_inner\n    self.run()\n  File \"/usr/lib/python3.12/threading.py\", line 1010, in run\n    self._target(*self._args, **self._kwargs)\n  File \"/\u003c\u003cPKGBUILDDIR\u003e\u003e/mistral/services/legacy_scheduler.py\", line 152, in _loop\n    random.Random().randint(0, self._random_delay * 1000) * 0.001\n    ^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^\n  File \"/usr/lib/python3.12/random.py\", line 336, in randint\n    return self.randrange(a, b+1)\n           ^^^^^^^^^^^^^^^^^^^^^^\n  File \"/usr/lib/python3.12/random.py\", line 312, in randrange\n    istop \u003d _index(stop)\n            ^^^^^^^^^^^^\nTypeError: \u0027float\u0027 object cannot be interpreted as an integer\n\nChange-Id: Ia85ea13923216d090ef1e4deee7840e70f9185b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/beac16281a2fec8ec6267a9ed82a1bf1d075abfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/beac16281a2fec8ec6267a9ed82a1bf1d075abfb"}]},"branch":"refs/heads/master"},"2b2c8759d4ef8632ffab1c5647ea3a2ff051e301":{"kind":"REWORK","_number":2,"created":"2024-03-19 08:05:02.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/73/906473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/906473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/906473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/906473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/906473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/906473/2"}}},"commit":{"parents":[{"commit":"08b5a9b7c617e7605f9048769af60ed94bebd825","subject":"Merge \"remove reference to old unittest2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08b5a9b7c617e7605f9048769af60ed94bebd825"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-24 09:09:10.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-03-19 08:04:43.000000000","tz":60},"subject":"Python 3.12: cast random delays as int","message":"Python 3.12: cast random delays as int\n\nThis patch fixes 128 unit test failures under Python 3.12 when\nbuilding the Debian package under Unstable. Without this patch,\nI\u0027m getting:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python3.12/threading.py\", line 1073, in _bootstrap_inner\n    self.run()\n  File \"/usr/lib/python3.12/threading.py\", line 1010, in run\n    self._target(*self._args, **self._kwargs)\n  File \"/\u003c\u003cPKGBUILDDIR\u003e\u003e/mistral/services/legacy_scheduler.py\", line 152, in _loop\n    random.Random().randint(0, self._random_delay * 1000) * 0.001\n    ^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^^\n  File \"/usr/lib/python3.12/random.py\", line 336, in randint\n    return self.randrange(a, b+1)\n           ^^^^^^^^^^^^^^^^^^^^^^\n  File \"/usr/lib/python3.12/random.py\", line 312, in randrange\n    istop \u003d _index(stop)\n            ^^^^^^^^^^^^\nTypeError: \u0027float\u0027 object cannot be interpreted as an integer\n\nChange-Id: Ia85ea13923216d090ef1e4deee7840e70f9185b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b2c8759d4ef8632ffab1c5647ea3a2ff051e301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2b2c8759d4ef8632ffab1c5647ea3a2ff051e301"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
