)]}'
{"id":"openstack%2Fmistral~906538","triplet_id":"openstack%2Fmistral~master~I03ccd5b7c7926c49c7e487d84aa379e4146009dc","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-07 22:16:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03ccd5b7c7926c49c7e487d84aa379e4146009dc","subject":"Remove policy.json file","status":"MERGED","created":"2024-01-24 16:30:27.000000000","updated":"2024-10-07 22:16:59.000000000","submitted":"2024-10-07 22:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906538","meta_rev_id":"21481996c7c91b3b5860be975de28e782c310878","_number":906538,"virtual_id_number":906538,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 22:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-07 21:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-07 21:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 18:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-07 21:25:36.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"a7f59f02cca245ef77655d7b8fe7420063ada7fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-24 16:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d6cbd3000fad019469b3693f5b0792b5bb25688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 18:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5034984ffa694ec084f1ccefc1c79b9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31182b9e9d7c42e9bbe56fe58293a586 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8590675cac8b47968bbaf2b5bae7eebd : SUCCESS in 14m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e08e92e004a456482a513559d055c5d : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9db9b1d0bc48f78a8f012f8bdbea91 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdf6d4632e45476ba78129e69739edce : SUCCESS in 16m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/86c28779bacf48b8844a15fa2aa97a25 : SUCCESS in 52m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f63d48e2b538438888cdd4ccd37b28ad : SUCCESS in 35m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b1a980303784afbb91ee2d169bab2e9 : SUCCESS in 43m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b2cb99225989482ab48afe3a9605c73f : SUCCESS in 51m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dacd251b9b6a4f2f9d99ac7de1c6d893 : SUCCESS in 44m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/52f63217ba244898b1323306ef61a8d1 : SUCCESS in 53m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/db8b0673a76f4d34a0136647c11462ff : SUCCESS in 46m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c3345e3d267a4fc6a099b35a6200dac6 : TIMED_OUT in 2h 02m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be0eaa710821872339ad343e65e64b0511d0c0bd","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-07 21:25:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"845e9ec20c504b56871f2198ab7c485d62b62454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 21:26:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f024572b6a93d43789523313666a060c75f750b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 22:16:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ef1f2839c0649c59699baa22ef3e159\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3fb301488dd487aa5b271c91d9f9a52 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18c2020f3db249aebd2166336e3d558b : SUCCESS in 9m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eebd89e228044e59a4d7f27a2d6193f3 : SUCCESS in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5e9e75974d4423923b1f8b2f888bd8 : SUCCESS in 4m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/70968a5dd3344299b67e9e2d1dfe0f98 : SUCCESS in 28m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f768b7e69734a59b530ee2a823e2c8e : SUCCESS in 30m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aae7b58f95684c68892646db68968839 : SUCCESS in 39m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6a215d0de52d494a956a2d54dc708081 : SUCCESS in 25m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e3a55df1017047cabc81e42f014833f3 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a143b3532dd3412150fb6057c6ad4899d6c6fc26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 22:16:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21481996c7c91b3b5860be975de28e782c310878","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 22:16:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb413496e084447bc770e293af5859c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a15e8f81b8641bca087006ed9412f2b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1323ed50015374a17cca7b4756b88ee2a096c499","revisions":{"1323ed50015374a17cca7b4756b88ee2a096c499":{"kind":"REWORK","_number":1,"created":"2024-01-24 16:30:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/906538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/38/906538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/906538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/906538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/906538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/38/906538/1"}}},"commit":{"parents":[{"commit":"b68c105696d9c5fe331ba3083b05bbf3c3db31ad","subject":"Revert \"bump eventlet to latest version that support python 3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b68c105696d9c5fe331ba3083b05bbf3c3db31ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-24 16:29:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-24 16:29:48.000000000","tz":540},"subject":"Remove policy.json file","message":"Remove policy.json file\n\nThe json format of policy file has been deprecated and now mistral\nloads policy.yaml by default. In addition the \"default\" policy file is\nno longer needed since all policy rules were implemented in code.\n\nChange-Id: I03ccd5b7c7926c49c7e487d84aa379e4146009dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1323ed50015374a17cca7b4756b88ee2a096c499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1323ed50015374a17cca7b4756b88ee2a096c499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
