)]}'
{"id":"openstack%2Fmistral~906595","triplet_id":"openstack%2Fmistral~master~Ife23f809674dfaaa83fb752c22d77bcc30aa044e","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-08-16 14:22:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ife23f809674dfaaa83fb752c22d77bcc30aa044e","subject":"Fix some race conditions in mistral CI","status":"MERGED","created":"2024-01-24 22:46:32.000000000","updated":"2024-08-16 17:29:50.000000000","submitted":"2024-08-16 17:28:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"906595","meta_rev_id":"7b794c643d58d595ea284653627ccfef599de46a","_number":906595,"virtual_id_number":906595,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-16 17:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2024-06-01 20:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-02 08:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-02 08:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 00:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 12:37:18.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-08-15 14:52:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"419e22e72000e24030d658262dcdeeebf611eb9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-24 22:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa97cc375fba81a35a13fc955758d94427d59223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 00:53:00.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9916792863ca42d98b65f260558d6971\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f75cba8e27e4e248ea6c0cc2e365c84 : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/747a7713e0614afa81aa2891b246fd13 : SUCCESS in 15m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a0fdefdf77744deac5a7ec584845a2b : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1152f4716b964e58a7a4f1f4536e5a7d : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f27d620d903402dbfc0157de6ee43d2 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a47d286669bf4509a4106b5b51255f05 : SUCCESS in 17m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6492f5b910f3420b963a7bd1fa08058a : SUCCESS in 36m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c5aeeeff4984944a298ad7c2f6aadeb : SUCCESS in 30m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17296345218e439badb7aa606ab9bf48 : SUCCESS in 34m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/72e71add7a364dcd8f6803b6a966a8f3 : SUCCESS in 42m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7f599887510a4759aca7f7b594ca299a : SUCCESS in 40m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5ccc9eb19a3a47f7b1bce28c462f20af : FAILURE in 44m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b160de75ce9e401e983bfdcfa2bae504 : FAILURE in 44m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ee8c64bbdca44cec97be2e85d5e58364 : TIMED_OUT in 2h 03m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0a89a31ce07ef310fd3a1ed8c4f9bb028bd56d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-25 07:13:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7da75658f34ddcbe1d11978ed009182ebcac44b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 09:17:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44bae43655a54ac69e02e325c53b5630\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3157fe1b972a43a79495a8923aad762c : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2b6bd5e73384508bbe5369a05f2bc90 : SUCCESS in 13m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b09c8828eec14b7a842cbea25ff6b8b4 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4332ed068434405810d6dcf17578749 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a348b17019245759d0f20db066b2e64 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c971dd5bc7144e7d87464cf6243cbb7f : SUCCESS in 16m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d696e647ad0c4643bbd2a2bcc957894d : SUCCESS in 36m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7af06cbc1c3847ebadc5e753c0924195 : SUCCESS in 53m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac042da3c77640129780f2bd0512af1b : SUCCESS in 34m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c7cb28ee4b1349e2b62dc9e429627305 : SUCCESS in 52m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b2255fe9fc824f9f8994becc89fa5488 : SUCCESS in 55m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bb25972dc8b24acc9db969037b298418 : SUCCESS in 50m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c8099575b2ed4a35ba1ad2b94d2e095e : SUCCESS in 53m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/75dde21ab6e14037b94905e4a481403d : TIMED_OUT in 2h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1d460ba9c542b0c5ab541aaa17219295978ca5a","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-01-25 12:37:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a6b42a03d706995fb66aef0a61f4a5009ae5945e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-25 18:28:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"386d2b6a587f78b364bc3bafaa1c11a9a103ecd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:30:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28875ea73b1d8431e799bc7075d437a841b741a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 19:29:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd4071a1e51948b29fd15163e3eed831\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f7cf8e4ba0489d8c8a56dbbfdd3956 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61b559b502e0458e89484e496d28ac52 : SUCCESS in 13m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a188b37061d44557955077b53dc4388f : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23730aa3479a43ad80a3beb06bea438d : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b6e4b7292a4ba7a1b3a5cf5eca4063 : SUCCESS in 5m 01s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/78846e25aa914bba9d95c66a13e534b4 : SUCCESS in 51m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19205e4df6a94b4c896be84001c9f433 : FAILURE in 38m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1090b6ecaa3044b288d2175d44d6e32e : SUCCESS in 45m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4b4b52b6ba5a4d80b0210bfbf8d0762d : SUCCESS in 41m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/88e344ff62dc4b259a8ae6071d513df6 : SUCCESS in 50m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dee50d28a25945639059015bf19ab0bb : SUCCESS in 47m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9c44771aec344b948b8ca0e5c8520fd6 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e1b0c9d76f28761ed07cba3c3cd945db67b51e44","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-25 22:22:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93c404454ec81cc7ea0d9de08f15f36243c186f4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-25 22:23:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d0de4cdffd47be5eda0efaec57cc791f231092e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 00:31:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/636fa6a209e4475d8b8006bc9196018e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1a1aaf3fe5400690a92e3f760fc6a1 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d56eddb7b8c4074998cab40f2bbbe31 : SUCCESS in 10m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e83460373894aa895c90e7be7fc4ea2 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78591ae4bb5f4f97a0a7bded939f1eb2 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f14161525df41e9b045fa829fff0498 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd7b983948714b57a27b65d5d75c6073 : SUCCESS in 20m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0ef523a1c9ea4004ad6891b3dbc6cfb3 : SUCCESS in 50m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06d46dcca8294e4896dde8c08b84fedc : SUCCESS in 27m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/758a0a9378604cc4b6b88288afbcad3d : SUCCESS in 52m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7785f437970545ee962a82eee8a04c33 : SUCCESS in 27m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/694f420eafb743da8d99466f930a7b8f : SUCCESS in 45m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d1c66d8583a64a25831a68007147443b : SUCCESS in 54m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/593c2081556747aa95fb639bbc14be5b : SUCCESS in 37m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d2a7106228a5465bbedb2bb95c94ba05 : TIMED_OUT in 2h 02m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18727b314939dbda0356c124e8e4ac9dec0ba28b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 00:32:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"217a0baf5fd6800aa5a318107bdc49af9cdcd226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 01:19:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8546665b48e34410b8a8097e5eee8890\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1263191b00814251910c9d597608457c : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec198135d0c644f7aa57f4d6db5222c3 : SUCCESS in 10m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c27def535e1b4a929512e8b2eeb89f24 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8580f8a545843c48b1a8c545c14dffd : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/929b4981870d464db6e0e24acf68f08b : SUCCESS in 3m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a51b06b518d74eeca7354d886eb8ec70 : SUCCESS in 26m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52fb779f839a48e28cc1873fe84eadbb : SUCCESS in 27m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0b9df75c5a04b86a78503c7b0983166 : SUCCESS in 28m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6e5af2753a8a472aac664e194d40e9eb : SUCCESS in 33m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e20ffb6149f44efeb48dd7d5d48b4655 : SUCCESS in 35m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f099ba03380a46cc9a9f85ad111679ec : SUCCESS in 45m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d395424efe734e378be346c90b0ca0c2 : FAILURE in 43m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c53ef10011616f6eb7955604f194f0e64c3ceb19","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-03-20 14:29:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"384e5509b90e03b6fc6b21b17c443afb3099c8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 16:36:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0818ca2173094fc29879275998b50d9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600898acd057456ab255800696dac66b : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b34a69f91ef49b0884dbb626712708f : SUCCESS in 11m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b380cb1a2274db8b1681eed958eb203 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf70b85b92c4450a8ddf3eda8a0d0cb : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/776ce16139bc4f51a3ce89fc9002fc67 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/702db15f0e224db7be369f0f5edc8970 : SUCCESS in 13m 38s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/729855eae3d64c778a8cd6c7221c8bef : SUCCESS in 45m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f976b8ec4f4457995002d77577070d3 : SUCCESS in 33m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e62f3c8c96414c0d818a8dedeb83b200 : SUCCESS in 49m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9e8dbe34defb43ff81d1f6c1ac08fdba : SUCCESS in 46m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1c3e829119004758a2ad19e68b4da978 : SUCCESS in 35m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/48c7e9abc5724872996c4560821c5b46 : SUCCESS in 54m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8fc516a52a9445ac84f2f6e840487303 : SUCCESS in 53m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2b0baf3e36234ae9ac65479218c46b68 : TIMED_OUT in 2h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1eb01eb65ec827e59af8e94c6757d9d7d762223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 16:36:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37eb525ff688660f614cd25c3e9728378dee6e62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 17:42:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f3583552eb401f8b50e09884ff03fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59152584ab634016a9275316fa3a8f0a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c96fd2222c8c4c458541f68b7dfe7167 : SUCCESS in 14m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b9d49c9502e48ea8d91b9bad08c144c : SUCCESS in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c51d9051d8b48e994bc9e2ea864b0c0 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7c764f2d48d40c9a28da460f80ff156 : SUCCESS in 2m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d4ef866d81a24f8fad74863f5aad698a : SUCCESS in 45m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ffe75ba1c50425c9f21d9cfeadaea94 : SUCCESS in 47m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d56947ba82a94275a81568dcde46d624 : SUCCESS in 44m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0b9c8d0183694a58ab03a5c9c5d4ff91 : SUCCESS in 40m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/226beacd403f4394a00cbb321e5e8d02 : SUCCESS in 55m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1757abc25cac408b994923f882f382e4 : SUCCESS in 1h 00m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38d20f8d29b04eaaa0195d279d462ebe : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":2},{"id":"548dcc73e25c98b9e93324522579bfa56f3c2bca","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-05-30 15:21:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28a74e4682d6630ca0609544c034fdb31562b90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 17:24:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3770b4770ada45d1911c0741de0354c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ff91c2e77a54da2a7e4228c5c72f06b : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da032c46af64137aa9c9973ad447db7 : SUCCESS in 12m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ee61fa4710647bc96fb4a0822278299 : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f3c0614072e47eba7ace686ab7dbce0 : FAILURE in 16m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0428b1d47cc3401a9c1e328d8c6e191d : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3253ed2413524b2a894df6412378a59b : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e204466e288244d7a686a3b04f926a8c : SUCCESS in 18m 51s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b3a3226d0be3498eb7b0688c18311f1d : FAILURE in 4m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2e5a3463ce134c7995aedd27ab9f7397 : FAILURE in 4m 14s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8383c33d783a4996977f5dee43ed698b : TIMED_OUT in 2h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58959fcc10562ce82e5bb4dc036435656b4bfe36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-06-01 20:27:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8458df784a54b4b7c6fbd6bbc90b6e3a5dc8149a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 22:31:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70005aa8a59d4c1182cfba0574226a14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d3548ae76e49429c688af9a3c4f59c : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cfe3c35090d492c80e77a4630b7cb71 : SUCCESS in 9m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69eb3064544649139bda0e2ab7ec79fa : SUCCESS in 8m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcf9fc9f57104179a92f8e3380b6b209 : FAILURE in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c1a79062494055a340b7e9993e4786 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/335cc73746a849b385e629bca3f48a83 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cb0fca48d5842328b4a8abacc83cc21 : SUCCESS in 11m 26s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/903ee42b9c044ef69ce771d46134cc3c : SUCCESS in 38m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fbd6d567e579465b965b700a57d959de : SUCCESS in 36m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4e589b431c5149d2a9ffb223dd076672 : TIMED_OUT in 2h 03m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e62aed34ced86a914a2c3b1a40f4b22542ac90b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-06-02 08:13:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a3b35cab7827c8d737fc43c29f0ea515b72558f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-02 08:14:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b60d4ea70d0d7e9140891a17b410e1ef46657df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-02 09:15:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/453008e54b96453080e3b7f1009c7fb2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3937405afa6467688503f9d0e2022a0 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97d03cbaa4ed4c50acabf28c3f81a74c : SUCCESS in 10m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f9f125933a045d6b2d4d9683fe400c0 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d17a0c74884fbd832de7a57fb83dc0 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65492e162522463d8e4d530ed4991845 : SUCCESS in 2m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/85b28a6e9976474aa68de208692b2037 : FAILURE in 53m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4bd77814ecc1456a937350908a51dfbc : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":3},{"id":"965c25111fe18433fb06b8b7198b36ab12b33e5d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 14:52:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"180a2b9c8375782f023f73ef2ad2b7c8c240b2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:01:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9acd5225c57c447594cd0c1c0627ce94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f0512affa0424b8a51311b7a89e52c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dee6d0c136047b99c4e89c124150f84 : SUCCESS in 13m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a59974999994feb8946be2e91de4be3 : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37e50232e3b14b2a982f4af36879ca0f : FAILURE in 14m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5f94253cac488e904fc4c222b1a2cc : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd3cff5ebb6b4876984cf74b7f2c0235 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6bfc34298854839bc9efe47fe25cfa0 : FAILURE in 16m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/023e2d4695cd47f897479b5585546b83 : SUCCESS in 29m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e88b7b2318b4b0abf13442231e42909 : SUCCESS in 39m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c50ca0e7b74247efb19d27db63acede8 : SUCCESS in 39m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/57408ccfa154455d949520822fcd68c5 : SUCCESS in 37m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/90cd7dfaab9143eab19c4e0e70a8e8fe : SUCCESS in 29m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/67cd4b0fa5db441b9ecf7ea4e6f2bf8a : SUCCESS in 47m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d04387a095ec4530be50a5b6b620f99a : SUCCESS in 43m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8f9905caa40c4cc0a9078caf197eb419 : TIMED_OUT in 2h 02m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8da3095075ed6d1827d6fb1f6cad96df166f7215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:02:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7011bae62fe7cad85e6282cd99dadd29a08efe12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:55:52.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89c0218b099545518994fb2f3cc04737\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae13d85843f4304aa1bfc1dd65fb7de : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b7cfecd065a44f6abe89ca72ec47ad4 : SUCCESS in 15m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fee13dfce02e494c87f4f22a7630628f : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb88fa6839d4125ab09902a5299a9dd : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb55a8d4ed1d4c258dd7b5a3232e34d2 : SUCCESS in 2m 52s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/49e0262c6e5c4c43bc00024d61e581d0 : SUCCESS in 36m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7af33bac7a5b4a11bf6dd65157c83700 : SUCCESS in 36m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d50f6941be0485780302611610d5793 : SUCCESS in 36m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2e3ab2132ea84829b33cb278f86aca2b : SUCCESS in 36m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6cd526224ae546589e93621d2b086e7c : FAILURE in 10m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0c845da9a11a4636a8d0937634b2dee9 : SUCCESS in 46m 34s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fb7d1d911b684ea8b30b6b135af88a8d : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":3},{"id":"5a6654ff33e4d345e14ec4992ef606e9fe9ef157","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-08-16 07:58:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40c6f3278ee4ba4da3d3a8fd8841e2b357b42fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 10:07:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/088dd3fe7842409e8b731a21a0f0968f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09004b8aecc44873839e1a9a28f90d53 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d85e955e614540f49fadfe3e7306539e : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7dbd16d569cb446bbde72dc48cf35290 : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a716a6d44637493f9b2648e1bc0341ba : FAILURE in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9235a0af1806417f9a6fe1bafcc37fa5 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29c6f3157d564e6ea83f2339a63e96b9 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/765e2da21cb44fe5a709f9676c69f1eb : SUCCESS in 17m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/799595ba92c54870b9ef5f676afe8438 : SUCCESS in 28m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eebc033d37649f68a01e76d87612de4 : SUCCESS in 25m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/476a151bf96c417fb18202dc365c90a8 : SUCCESS in 26m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9c0eefa2ddcc49baa86d5b87d81a16f8 : FAILURE in 38m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/771b1c9733a74402a8ddba32b88ad1c1 : SUCCESS in 39m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5d781f166a2346509f298e82ec8fa5f1 : SUCCESS in 55m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1f44436932a8415c957414a485f13421 : SUCCESS in 52m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7edb240762f84351a2fc5d1fff8e03e8 : TIMED_OUT in 2h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e49984013a70112c0172b8413a97dca65568617","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-08-16 14:22:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97210546a92adf622b917625a801f64d9880272f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 16:29:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7aa2aaf4864b77a8bd8f0217573eca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b33571d34248c8acea2e7571f0eade : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e2e1087b4c465b82eaca8b08e3b021 : SUCCESS in 13m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd69e36e4a4c4878b5740eea4c6954d6 : SUCCESS in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d10bbeac62894768acba91fd144795d9 : FAILURE in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158c305b9aa34c1c893bd4d7f18b26bb : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4117f0d76e74b35818dae4455295288 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08ee7ca85b5d47cd8d0021d7ce3e67f3 : SUCCESS in 17m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ad329bae893446079729ecfdbb8dd132 : SUCCESS in 37m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94fd56c5d0ab458eb11fcad39dc1b8c7 : SUCCESS in 30m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d7fd13f7274f478eaf55d3cd7ce617 : SUCCESS in 32m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dc93c8c5e8fc4ffcb4eafba336cdf9f4 : SUCCESS in 33m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c4336bf3ef294f7b83d881fb982fff8f : SUCCESS in 28m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/73f38bc3979048ada80093d256b6371d : SUCCESS in 49m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dafcb113d83145479a484e92b995dc3c : SUCCESS in 42m 24s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bb60e8a7be0b43cd9b9fe6ccfbbf8e45 : TIMED_OUT in 2h 03m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b3cb5dc734cdc3425dbf87ae884875e6e39a9b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 16:29:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b6d30b3e81a09956f676360ee86b341f61f6a699","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 17:28:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183a9e6870494f6693d0f0948b3cf921\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7ffcb888f4449facdd3e3db7853c65 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d1f69d1b80e4c1680fee72a7b7dccf5 : SUCCESS in 9m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea1bb4229a89431fb93b343be297c22d : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f33e3aeafab4b659c2181c9d98b5961 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/767913dac0df412ab317309ec9878cf5 : SUCCESS in 3m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b87c2fd3b07740029a7c9c476e5e2a47 : SUCCESS in 27m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba2281e61ca54cc3b424b2a471fb9f5a : SUCCESS in 27m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a551a2f136ab4477aec528c467a032b7 : SUCCESS in 25m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6f45525415014a3db5e3c4eb80ac5b79 : SUCCESS in 26m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/77b0723754b64c65a7266777056bb686 : SUCCESS in 39m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5a872742a4574778b84d93119be4e49e : SUCCESS in 57m 32s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e7e6980277914131a2803dc347873ef7 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":3},{"id":"5111b5ffadaa249660467823d40ad902d164b8bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 17:28:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7b794c643d58d595ea284653627ccfef599de46a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 17:29:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668b2bc1b7a94e5c98f8c5cabce2f566\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9f87af906e401fa628efc74faeb518 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4429e5a2e379441db5bac5c343c1695b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3e8f5b0437c2497b1f5e80e50fb873df38f9f53","revisions":{"e0a252bb02605f8d3453a4ed223a0bcbcdecb93d":{"kind":"REWORK","_number":1,"created":"2024-01-24 22:46:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/906595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/906595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/906595/1"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-24 22:45:42.000000000","tz":60},"subject":"Fix some race conditions in mistral CI","message":"Fix some race conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n\nBonneNuitLesPetits\n\nChange-Id: Ife23f809674dfaaa83fb752c22d77bcc30aa044e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0a252bb02605f8d3453a4ed223a0bcbcdecb93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e0a252bb02605f8d3453a4ed223a0bcbcdecb93d"}]},"branch":"refs/heads/master"},"9f783ae6e935e247a3bceac9cbb38b279b22ea94":{"kind":"REWORK","_number":2,"created":"2024-01-25 07:13:41.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/906595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/906595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/906595/2"}}},"commit":{"parents":[{"commit":"0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510","subject":"Move zake to test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b7c4cd0f6c52e1842060c2c65cd03c88d0e8510"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-25 07:13:35.000000000","tz":60},"subject":"Fix some race conditions in mistral CI","message":"Fix some race conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n\nBonneNuitLesPetits\n\nChange-Id: Ife23f809674dfaaa83fb752c22d77bcc30aa044e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f783ae6e935e247a3bceac9cbb38b279b22ea94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9f783ae6e935e247a3bceac9cbb38b279b22ea94"}]},"branch":"refs/heads/master"},"a3e8f5b0437c2497b1f5e80e50fb873df38f9f53":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-01 20:27:03.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/95/906595/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/906595/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/906595/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/906595/3"}}},"commit":{"parents":[{"commit":"0e03f2db5ea144e5ea840cd31ba26d1061df869e","subject":"Replace deprecated assertDictContainsSubset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e03f2db5ea144e5ea840cd31ba26d1061df869e"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-01-23 23:37:47.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2024-06-01 20:27:03.000000000","tz":0},"subject":"Fix some race conditions in mistral CI","message":"Fix some race conditions in mistral CI\n\nWe recently had two issues in CI, this commit tries to fix them.\n\nCloses-Bug: #2051040\n\nChange-Id: I05d850344c653bc858b4565673606531da803342\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n\nBonneNuitLesPetits\n\nChange-Id: Ife23f809674dfaaa83fb752c22d77bcc30aa044e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3e8f5b0437c2497b1f5e80e50fb873df38f9f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3e8f5b0437c2497b1f5e80e50fb873df38f9f53"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
