)]}'
{"id":"openstack%2Fmistral~913334","triplet_id":"openstack%2Fmistral~master~Ica41e8208b29aeba6e781801540843d46375c3d1","project":"openstack/mistral","branch":"master","topic":"bug/2058009","attention_set":{},"removed_from_attention_set":{"29124":{"account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"last_update":"2024-04-09 06:29:17.000000000","reason":"\u003cGERRIT_ACCOUNT_29124\u003e replied on the change","reason_account":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},"36440":{"account":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"last_update":"2024-04-09 07:43:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica41e8208b29aeba6e781801540843d46375c3d1","subject":"Fix maintence mode","status":"MERGED","created":"2024-03-15 13:53:29.000000000","updated":"2024-04-09 07:44:40.000000000","submitted":"2024-04-09 07:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913334-bug/2058009","meta_rev_id":"2282adc1e848128d3ead633de86c5ef2507b2ec3","_number":913334,"virtual_id_number":913334,"owner":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-09 07:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2024-04-09 06:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2024-04-09 06:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 15:51:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2024-03-21 15:11:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"e858e3be5f0e8c440f9515010af4d3285a570d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-15 13:53:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8aacab6fb43bdcbfbdf9b29010f2baf243c01c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-15 14:44:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b4194560c14149676dd1738896014c67e5df72e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:51:08.000000000","message":"Patch Set 2: Verified-1\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79b545893fe145c8bbe9999563ce1b3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb4f195dd6d44339538326c451a1373 : FAILURE in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4030fe74ade34fe090069327f242edc0 : FAILURE in 12m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee07a901445647d48dc123a5c007898e : FAILURE in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825d54957b3e41d4955f4d968dc68e3c : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/685f62c54b414f71bea48a94682710bf : FAILURE in 20m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/14b029135e904422b11221f1d8f29435 : FAILURE in 39m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dfbfea2140141d7888f52b17826fb04 : FAILURE in 35m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5310095c3c34b64a76607d7e2ca6c4a : FAILURE in 42m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ef860ebde5ce48908266a2a29cda0ff7 : FAILURE in 32m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1295f8285ee444f789057277ea886290 : FAILURE in 34m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/076da65a64d94724bcc659bba25da5f2 : FAILURE in 47m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a1fe75e0690a42b0ac5b7641271ca30a : FAILURE in 44m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/884281eed4cd48c89cf7c89b07d71f03 : FAILURE in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"501cf32ca6fd367760d5fa540f9273a8d0940ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-15 16:02:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1cd03608de9ba82c87c06d49630560df06eb94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 17:58:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9584bae067e545f4844cf94195f315c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8354cb38c2b2467cbd3ff234733c8085 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7938aee3d046c38d00309bead1e767 : FAILURE in 13m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec47755a3316443baae4f4fc6ae7e9fc : FAILURE in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41621f830942440eb6f495f201acadc9 : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7183e68d852e40e99d48c7c4c3354ad4 : FAILURE in 18m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f45f5dc303ff46dfb1c6b25894540612 : FAILURE in 47m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ecd2412117c4c069b986f2b70494eb6 : FAILURE in 30m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbf1f32218ea4bc5b9c7d927283ae579 : FAILURE in 43m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3121c65be3b4429299a230a8595e0512 : FAILURE in 1h 46m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3e3b30a1d81e4ab395832130a233238e : FAILURE in 1h 26m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5f5460d158f649b7b4caecaaf9542bd3 : SUCCESS in 42m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/db1e6c81575941a482a76ac849fc82f2 : SUCCESS in 53m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/442731ccf11842aea55cc06d8c891527 : FAILURE in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd9cf5240541353903751c5693195679d25be223","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-15 20:27:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c67559ebe6b462f3e50bf2a58bdb48d135bec1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 21:26:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80d70469333e4a5da624ff838d9fe0f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053803652f4f4be2a7bd4dc784dbe8be : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0765f484e2c646c4b576aa08d93f69eb : FAILURE in 14m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3406aab4dce04d0592e7ce85c2a781c6 : FAILURE in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3678b8c9f93c4e6dbc74c6027fc4bffd : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/210d538a2d2e46ef9eb60d12278819f8 : FAILURE in 14m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2517c98aa707488296afdc5e47234862 : FAILURE in 34m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3df40bd819d4972bd54d3cbde12478f : FAILURE in 33m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e898fd154658460d8f94ed415fbb435a : FAILURE in 33m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/834c7858d3504dd88b50c9a4e707f61a : FAILURE in 34m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9979577215f542dcb465a78760453ff3 : FAILURE in 31m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b715bfd7474546d288a4fbb0e85763f3 : SUCCESS in 45m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6af48fae8a8a445483a9aea658f72716 : SUCCESS in 52m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/de8d7d1d5ae749bda189bcfdfdda9cf1 : FAILURE in 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"131e28f88dfb5b2d961b073875365989a1fbd0a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-18 08:56:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"931694188645ba0bd67a762d2c10390c085b2b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 09:47:39.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1cf3e201061483291f6b7f8976cb32d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a602099f2ac84a7db6d606bdf5be5d50 : FAILURE in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16942749b7ee45889c58c2852ed80f50 : FAILURE in 14m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2aec3b119a304cad869083ca7b1f83bc : FAILURE in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfd22826b0445e6b9a6aaa2bd26e851 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae171fd9e7234317bf65aeb7b5815d3a : FAILURE in 12m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8234814988fc4ebc92e076f95dfea446 : FAILURE in 34m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5fd3b999f9d4e49ade55540ef01048f : FAILURE in 35m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d94672a7c0b64935891f3368b0e09159 : FAILURE in 39m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/118eec56e23d44b9b1d885a33ce1aa15 : FAILURE in 44m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/94af631060604e2085d365772eaf09ba : FAILURE in 36m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/37649df91e544b9eb4661cd839c3d495 : FAILURE in 47m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/952f81a151e94ca886c4dde455255dc2 : FAILURE in 41m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a93b26c5ce14400cafcf5480815a0dd9 : FAILURE in 36m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1fc19f37c999e0666349ea7a27a36ddb7a282136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-18 10:29:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0fa571713dc4dd82a726b01dfcb80532bebda6b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 11:21:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/425394e8247b46ed8acdebd3ef47d6e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007f8b026616458db8db5f820add11d5 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b18c4e62226744fda422bf95c47e3c7b : FAILURE in 13m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8915eedb6fc42bfb2e01610cbc6c0a5 : FAILURE in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8331f0e5f51a475c96b9cdb187294176 : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc1d98e7f3d4ca5977a31079497c3b1 : FAILURE in 19m 17s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c884222d993040d598195fadc6d5979c : FAILURE in 32m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c806e53b570e4ff4b1332a70756ffc0d : FAILURE in 34m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a7c5344de5d42da9e8106d6bff38533 : FAILURE in 45m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d141b4f44c534a18808324a48d9ed090 : FAILURE in 42m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a8986c57b7ac41909b505e1fe1d762ee : FAILURE in 40m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b0ebefeb622e4c4096a5930a758da17e : SUCCESS in 47m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c9f12ba73eb24ce99382f60815148dbf : SUCCESS in 45m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bee5388fc23b4e2c9ef9e4f19645d7fa : FAILURE in 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f47c7043b13a8c2e2abb9b505c0d7af70f0d1786","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-18 14:34:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5109954f813467a85d347cdab4d36b53664e42e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:28:16.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb6de6ffef924ad08aa063b865915deb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e7e4a2bafe47fc8f74e8ba10c4ce11 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/789ab0c59fb74da6b6199d46b6909ffb : FAILURE in 13m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90801becafdb4ed0b2d36980df57347c : FAILURE in 14m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc51bb3f7af46e5ad86af4450de0c35 : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ba03098b7f14fa5867946668199711a : FAILURE in 16m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ffbc0cc947dd4a809646d59ebe401e87 : FAILURE in 45m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/534e64c802584114a0862ccc5eb3fd47 : FAILURE in 32m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b878f5e655084ff3b5bf505fd0617b71 : FAILURE in 39m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1d51d11a214f43a695d839e6541886a3 : FAILURE in 33m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f8a9228d44d64e7d878311e9e2cbd8aa : FAILURE in 46m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/205c1484748241f6a02e609504651115 : SUCCESS in 43m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b0c9f4bc5cd444b8a2afb7298726d486 : SUCCESS in 42m 24s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/cedfb20b2ac042d9973140154162724a : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e5b4f31298caa4d8fec72e7139051674b5649b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-19 08:39:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"87553b153259049084b0e31cf1ae4db976f509f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 09:41:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/036667bbd40a4a209b6c9339db5c677e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d088233d8314e4bbad9d693cfae5a70 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38f190d4fcce41c2b5216edbb4ef061e : FAILURE in 12m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aed6290b003744849ed0ad5fe31692e6 : FAILURE in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcaafc5e0c4246a0a030a1674deba38e : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e29a74bc09c4210b4ee10af38852cdb : FAILURE in 19m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b70827438dfc44f4ad46677ef2dc5d35 : FAILURE in 47m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc1db033b644b5ca68dcac87896c982 : FAILURE in 41m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d029da1319c8438b9bd6fdcf46fd02b8 : FAILURE in 42m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ee7b4733dfc54c468db100d0f2b071a0 : FAILURE in 45m 59s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/00c9419b081a4048aebfaf22cd64d1b9 : FAILURE in 43m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7d2a46b5935a47a190455ffb2ee18489 : SUCCESS in 53m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5ab8e7ed5b004caaa74077c0e1d1ba19 : SUCCESS in 52m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a2053bfd17f14e31ad840f8b361f62aa : FAILURE in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3f5c25fa99b800ef6106418f98bfd2106358fea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-19 20:42:06.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8fdbbe63335cf3daf267451b6033091827c7debf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 22:50:57.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/923e0a2fd7fd4408b593cd00a1ca9695\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2e344941b741fcad630ae85af39656 : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c3231f6409f47b292b808c5d48aaf8d : FAILURE in 10m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b85ac9ca3a64b58b680144f044e942f : FAILURE in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8910c08a336d4bf983f1753107f6fe91 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816f762089d6405ba959b53824bcd823 : FAILURE in 12m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73d60201341b4e0a8e9e449ca136fbc0 : SUCCESS in 35m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/175523d56ff24da6a95f3c29fa68e985 : SUCCESS in 28m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc31c0ca5e3a42f8831ccdf1016af1f8 : SUCCESS in 35m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/417453e953d047fcbbdc5e42ed580748 : SUCCESS in 26m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/91a4a9c73bb84f5483302d2432448a41 : SUCCESS in 44m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/51a275408eeb4972a3c56ae128ab2751 : FAILURE in 54m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c4b0ee824e9848e78b2bbadca0cd4a84 : FAILURE in 36m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/49917be051c44053b1d9ea506e73b05b : TIMED_OUT in 2h 02m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"34e8a33760d898caa9154a43ace9a0c94126c004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-20 07:00:51.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"41b8107d5b7334966ad7ea3f223a4374dbb1b02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 09:04:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d0ce98b8b74454081dffdeb72f16025\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa6797a714b461e92137f8db3b9b62b : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e738865dbf4773b8f268a13f248bdc : FAILURE in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e58b4628e6884345ad61373100c90cbb : FAILURE in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664d392c2b4d4de5bf1e4c132b2c88df : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ea4ccef5f334398980e7d0b58389063 : FAILURE in 12m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/07cc86d69f094e41ad352c8da3208e20 : SUCCESS in 47m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4394b73edf3246f588da6ef9079c104d : SUCCESS in 28m 54s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/840f1838dd1e4a89b8e4fb81a7045cfe : SUCCESS in 46m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/804c3bec515f4f69a9d5523241999eb5 : SUCCESS in 46m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f4ab76157da946179b2550e19ac8f2ef : SUCCESS in 47m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/98e12b6ed9ea4daf94d44d52aa89284f : SUCCESS in 41m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a795121ec7ba46ca8691cc38fa8b1b40 : SUCCESS in 52m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/57e253c0850d46f5899ce444db3fa508 : TIMED_OUT in 2h 02m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8bed743ff967779911e5dbed2432dd7fb725402f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-20 10:51:21.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"a14746a3a98ae576be5128b7f93ee51c22f09b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-20 10:54:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f46711d23e589cc95b17de4452a311e87b6fbdfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 12:58:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c6780f0e6eb41e4a60c58f760b8b4f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597d7afe171240a99c18cc7f60437436 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21338324653645069fb0babfd915cadb : FAILURE in 13m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/622f49ed4f184f3eba7f4367e9230116 : FAILURE in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889c3b4f0a5e49e7b38316f31aed1854 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c6c1ae00c55463d8238deaf37b35cee : FAILURE in 17m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/72e9dfa19340452686a509a7f254db43 : SUCCESS in 40m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96fcc8bf4dfb483f9bf671c06d91fe09 : SUCCESS in 39m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b28fa0415394374b59c6918e2fb6d37 : SUCCESS in 44m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/88d5d533724948e38758ae714d964752 : SUCCESS in 31m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0e7dde24fa93497c851278a1fee4c810 : SUCCESS in 34m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3a9793deaff940c9b163e334a0642366 : SUCCESS in 43m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a958b27473944286918855fed950fcff : SUCCESS in 48m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ab01f3f59ec4424185949d0053708605 : TIMED_OUT in 2h 02m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"21d7473e869d4f3167f8e99aa8da2f0b4dbed325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-20 14:05:19.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a2ca9019c0a195963ee8640304d38b3244f42d4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 16:13:47.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7f192f388f046df82d2af868fd2a2c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794a0a3e5b0642c69a603d82b8bef861 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7d600d852642289c78fda9fbf9683f : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b7799aa06e14f62996fbefc2354bd78 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9011f6533ea47eba422948c4e3060d9 : SUCCESS in 8m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e6678e23a244a8c99d0f090f2ce423f : SUCCESS in 16m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4397fb9938e141399600003cd8d54dc6 : SUCCESS in 39m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81ddd656f3c24c9b80986b18e8e5820e : SUCCESS in 51m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e5f2ba76a974560924578be17fb9a65 : SUCCESS in 53m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a6148ec5cc434e4887f5c974d7afa556 : SUCCESS in 42m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6fc0b42b383b4e03bb0b36598c427ce1 : FAILURE in 55m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/91d408d019e54ee9a334d1f66ac35fac : SUCCESS in 55m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/62243a4980044cf78ded002649d8e1cf : SUCCESS in 39m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/966babf08e60499aa98f037730e16549 : TIMED_OUT in 2h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8ebb6b9ae9d262d93a51bb4b727843f5ddf57ef7","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-20 16:29:22.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b7264aa1ade447cb973078bdc82ea07995299608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 18:44:08.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/757d9c37fafb4e1ebaef389d6bdeb439\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b96450e5954620bcfc39e29943221d : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5061a6c902164c1aab03d9ff885608fd : SUCCESS in 12m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/699c2f64beac4276834551f1de984adb : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafa5a68657446c59298ca4e443fa75f : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0f3f358d7a249419a1798743dbad843 : SUCCESS in 19m 12s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/74ae22f3647b48d5b43b950ecd87e5d0 : SUCCESS in 43m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/632a8379dc1c45c19da7901eeb0f2a22 : SUCCESS in 35m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/152aa339da2345cb9782cd719e3c42cf : SUCCESS in 53m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/38c06f1c1aff4ff987382529e965674d : SUCCESS in 36m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e537d0701d1e4207922334ed1082e65c : SUCCESS in 40m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5f62712536f14b1ea415d0ef32393bf3 : FAILURE in 42m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/33e6a48fd9564495a2dd0743b04e56b3 : SUCCESS in 42m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2ad1077360614bddb77eb8fcf5ca302a : TIMED_OUT in 2h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3f3f88bd9af0c6baf04007d67d29c074980fa5f1","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-20 19:18:38.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c5b0436183bb1bb94a0594471b9076f201ec8bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 21:22:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d476f4878894a05bda6460128cd9049\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea54e8aac2d484ca8a22d6087df2ea9 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/679925e2f5104c5c9bfb49733bafee91 : SUCCESS in 9m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df4080c9cb0a418fabcc42388d2928dd : SUCCESS in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c327b192ac4aa8a1c0a506e09df02c : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7a11864e6074c34b7ba874eafeaf58a : SUCCESS in 12m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ce01234765974b479ecc2359df9097ff : SUCCESS in 30m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dbad3cf2ec948f1ae8cd224642c65a0 : SUCCESS in 47m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/151478c2b0f8455c8f8c53f6a64a22dc : SUCCESS in 45m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/47b225af671f45a2b25869c3d46368ea : SUCCESS in 45m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/772471f2f18d42a3975785991cd280a4 : SUCCESS in 28m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/98e1c6c831174094b78e6d95e251256b : SUCCESS in 55m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/58201875d0cc411f87394a7569aaf5ea : SUCCESS in 54m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f0adc9ca6f8c4acc9ee9bf539203f58f : TIMED_OUT in 2h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"040ad6981933cad7096ac333d25aabbc30a90431","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2024-03-21 15:11:00.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"4834ea11e333c72bf2dad61fcb43ab5bad3fd72b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-26 10:51:39.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"01cbe8dd121f645c1dcc55b389b7eb1b427ce480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:03:21.000000000","message":"Patch Set 13: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffcbe6b3a21f45c481d692ebfa3c92dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189179db60114c2f92a87d7f88239979 : FAILURE in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb37782a06f44fadb5858f4acba05739 : SUCCESS in 12m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8e805093ef64df797744ca425065b08 : SUCCESS in 12m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3ec4a2c63864622b4d2f1a580f714ff : FAILURE in 16m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2dd4185222442d9ce1d88b39c3c287 : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c447d3f789e49b380a7a3aaecc9a113 : SUCCESS in 16m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5330d5c6eb384536b522f48aacaa2cbf : SUCCESS in 53m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/311c4c1a4b664ef6a18004a1e30c9ece : SUCCESS in 48m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da87b777a4c041bd99112894d41b6573 : SUCCESS in 41m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/73c2113c68264d10944ef40ce451a032 : SUCCESS in 33m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e9c88eb466ff4827925bf50d42aace97 : SUCCESS in 39m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bb05e5901a65494590093c8199db9115 : FAILURE in 45m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f1c1d3cfe77948a1b80c699f5b61215a : FAILURE in 39m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8e4d319e65574e29bfd7dc2fb26c3dce : TIMED_OUT in 2h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"af2e36c7fb2ac151cab799710c46b184aebdd5d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-03-26 13:32:29.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"364ba0a21a9914848cfff320d9ee80e5945f5187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 15:39:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68efbb63b43f4747b9717c7d363375b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f314fca17d8d41948bbc15a7340f5e71 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1792cc32f304e038e0f079a1d071761 : SUCCESS in 12m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ac8fa076b844a8e9a17fa68eefffe03 : SUCCESS in 10m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e235fef04df4a4fa5a924df59512d47 : FAILURE in 16m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18133d7b7a1945e2bf47c173a158baff : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a6d8cdf85f845f984899bf53335fbb2 : SUCCESS in 15m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3ef698e09da34334873987a453ac10c5 : SUCCESS in 34m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a4d5be49e2e461ca6d80fe48dde1042 : SUCCESS in 40m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e55b3089cf0341e6a9716dd4fd3a2ed3 : SUCCESS in 38m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e3167b8f0ac04dbeb749cd5a704ebcc6 : SUCCESS in 40m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/876c04ffa28f4788bf01a483e85a30e8 : SUCCESS in 40m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/880ecba8450847c6beec59488fdde3b1 : SUCCESS in 53m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dffb6f89ddc74e07a869bf2c573a0acc : SUCCESS in 40m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b9d5a774089b4ab9af33ea0bdb7d57cd : TIMED_OUT in 2h 04m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"06f06b5712ef5e41dd01703d5007315a9a4efc65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-04-08 06:50:51.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"c008fdeb2777517a6e4180abe6ca5599ef8098a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"date":"2024-04-08 07:24:41.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"41351adae95d2675f0b2715c0d7f6e3f7d0b2191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 09:32:20.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f252853789d94105bf583f9ebb28d666\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957e059d777d418cb315c6f0bb0933e0 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61dda13dd43342c3a7927b4be4cdbf94 : SUCCESS in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef2e8e3d8aec4a38ae400d91d3212fed : SUCCESS in 12m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1687acfb23984736890268083fcd3984 : FAILURE in 17m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5bb5a03e29421b911b3a3e02b34e2f : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aef81362c31842449103a4c1bf9f2415 : FAILURE in 16m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d4786772ecd249d5a4e060befee12aa6 : SUCCESS in 49m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bb2a9fc1df9463f9d0ba42f618bdb82 : SUCCESS in 47m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c746a1e04bca4799a8f394c377e77e8d : SUCCESS in 35m 48s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/658262825cc34da0abf11dddd4a7a0ff : SUCCESS in 50m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5fde27acb2fc46e69ba6d848d27bf3f9 : SUCCESS in 37m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8d43e3d9b2b648b1b0299d683d91e9c6 : SUCCESS in 55m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d3fba0bef16940cf9f39e9948c94760c : SUCCESS in 53m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d19ef65f97944b618f11acd9f1e0ccb4 : TIMED_OUT in 2h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9491a61bc02ba88c120dcc0a56fe664512194065","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2024-04-09 06:29:17.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"160be92a7caaa6d0714c4119de9505e2ae62d790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 06:29:50.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"7cc036277c31bb338bfd6776bcbeb1c7f9a0b732","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 07:43:43.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31d40b84993b4bf28bfc1e0933b6bc31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221123eac1fe45509e5151b32e826f07 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/804ecc8d6cd443c58aa94d515513bdba : SUCCESS in 13m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e444054f2432463fbea60e17a292d966 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9746c38cb5f04ed196d38ebebc6f6097 : SUCCESS in 6m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2d36140b2bb245feae3106772f5afa3b : SUCCESS in 47m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a268ec6d73941c08f19c166b0172c73 : SUCCESS in 28m 44s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85f5df76a76a4e6883caff4254ed9c86 : SUCCESS in 47m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d26019534b6b474e8633cd60cdce9b4d : SUCCESS in 28m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d3e13840ec954e9782952659784befd4 : SUCCESS in 46m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5d3424a37058427a886fb91307f07e44 : SUCCESS in 55m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3536966430ec43148ea13bf21b40351a : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":16},{"id":"9eacd13d6d2b4ca3b2f93b0fc35308b032b50284","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 07:43:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"2282adc1e848128d3ead633de86c5ef2507b2ec3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 07:44:40.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda8ce1b470e4ea88ca75504c3b384fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b45ec5aa184f9a920cc0dfdc2bd1fd : SUCCESS in 38s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"cadfe9dca2490afc05d2c2489db700a417cd0f0d","revisions":{"84bbac31fcbd14f0f69fe5515febb020e3c07b33":{"kind":"REWORK","_number":1,"created":"2024-03-15 13:53:29.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/1"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 13:53:15.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84bbac31fcbd14f0f69fe5515febb020e3c07b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84bbac31fcbd14f0f69fe5515febb020e3c07b33"}]},"branch":"refs/heads/master"},"2aabaae5e4b4c29797173601e1e09d3cbfdb69be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-15 14:44:50.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/2"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Zelenevskii Vadim","email":"wortellen@gmail.com","date":"2024-03-15 14:44:50.000000000","tz":0},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2aabaae5e4b4c29797173601e1e09d3cbfdb69be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2aabaae5e4b4c29797173601e1e09d3cbfdb69be"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eb8bfd710aa49c3edc3d1074172cbb062e31788b":{"kind":"REWORK","_number":3,"created":"2024-03-15 16:02:38.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/3"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 16:01:52.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eb8bfd710aa49c3edc3d1074172cbb062e31788b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eb8bfd710aa49c3edc3d1074172cbb062e31788b"}]},"branch":"refs/heads/master"},"0e5b6cd898ca614a26bc9549c428a00503ddde99":{"kind":"REWORK","_number":4,"created":"2024-03-15 20:27:57.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/4"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 20:27:41.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e5b6cd898ca614a26bc9549c428a00503ddde99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e5b6cd898ca614a26bc9549c428a00503ddde99"}]},"branch":"refs/heads/master"},"5b30e70715773b2b28ce7828cf8547506415310a":{"kind":"REWORK","_number":5,"created":"2024-03-18 08:56:36.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/5"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-18 08:29:26.000000000","tz":180},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b30e70715773b2b28ce7828cf8547506415310a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b30e70715773b2b28ce7828cf8547506415310a"}]},"branch":"refs/heads/master"},"04e01a4dab2286dbbcbd1adf4b29d6f531779aad":{"kind":"REWORK","_number":6,"created":"2024-03-18 10:29:48.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/6"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-18 10:29:16.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/04e01a4dab2286dbbcbd1adf4b29d6f531779aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/04e01a4dab2286dbbcbd1adf4b29d6f531779aad"}]},"branch":"refs/heads/master"},"66e1278007d69c5cd171cf7073c17dd43a779083":{"kind":"REWORK","_number":7,"created":"2024-03-19 08:39:49.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/7"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-15 12:29:31.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-18 16:54:34.000000000","tz":180},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66e1278007d69c5cd171cf7073c17dd43a779083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66e1278007d69c5cd171cf7073c17dd43a779083"}]},"branch":"refs/heads/master"},"ebd1dd1bd0744fc02f464aa31f653e8829c9b321":{"kind":"REWORK","_number":8,"created":"2024-03-19 20:42:06.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/8"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-19 20:41:18.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-19 20:41:18.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ebd1dd1bd0744fc02f464aa31f653e8829c9b321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ebd1dd1bd0744fc02f464aa31f653e8829c9b321"}]},"branch":"refs/heads/master"},"9250228f9a964a7bd4048f4c989e955bf84d6d16":{"kind":"REWORK","_number":9,"created":"2024-03-20 07:00:51.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/9"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-19 20:41:18.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 06:58:47.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9250228f9a964a7bd4048f4c989e955bf84d6d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9250228f9a964a7bd4048f4c989e955bf84d6d16"}]},"branch":"refs/heads/master"},"3be1dbd0be829e98daa91571b3296cd57c5d9f33":{"kind":"REWORK","_number":10,"created":"2024-03-20 10:51:21.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/10"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 10:51:06.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 10:51:06.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3be1dbd0be829e98daa91571b3296cd57c5d9f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3be1dbd0be829e98daa91571b3296cd57c5d9f33"}]},"branch":"refs/heads/master"},"6c9dc43a2fd3a50de829a7da7cd2ac451de84a91":{"kind":"REWORK","_number":11,"created":"2024-03-20 10:54:32.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/11"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 10:51:06.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 10:52:43.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6c9dc43a2fd3a50de829a7da7cd2ac451de84a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6c9dc43a2fd3a50de829a7da7cd2ac451de84a91"}]},"branch":"refs/heads/master"},"650fdbd58bcdf48ed43afea611b0ae6e39db8215":{"kind":"REWORK","_number":12,"created":"2024-03-20 14:05:19.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/12"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 10:51:06.000000000","tz":300},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-20 14:04:24.000000000","tz":300},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/650fdbd58bcdf48ed43afea611b0ae6e39db8215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/650fdbd58bcdf48ed43afea611b0ae6e39db8215"}]},"branch":"refs/heads/master"},"c46ef74583b37551a87f2963b1645d6c49d76a4a":{"kind":"REWORK","_number":13,"created":"2024-03-26 10:51:39.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/13"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-22 19:11:12.000000000","tz":180},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-22 19:11:12.000000000","tz":180},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c46ef74583b37551a87f2963b1645d6c49d76a4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c46ef74583b37551a87f2963b1645d6c49d76a4a"}]},"branch":"refs/heads/master"},"a52ff249505eb1f8d26fde7060a230c8973eb90c":{"kind":"REWORK","_number":14,"created":"2024-03-26 13:32:29.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/14"}}},"commit":{"parents":[{"commit":"14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b","subject":"Merge \"Add back wsgi script generated by setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/14f40d9a5134b2837e7f1cf82e0cfafcd0ae604b"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-22 19:11:12.000000000","tz":180},"committer":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-22 22:01:35.000000000","tz":180},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a52ff249505eb1f8d26fde7060a230c8973eb90c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a52ff249505eb1f8d26fde7060a230c8973eb90c"}]},"branch":"refs/heads/master"},"babb381bfe1a58be337aeed15c5e7ea6c248c3da":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-04-08 06:50:51.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/15"}}},"commit":{"parents":[{"commit":"4446f9ced80caf6bf81fcd2b0ce0fd9f336b7588","subject":"Merge \"Python 3.12: cast random delays as int\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4446f9ced80caf6bf81fcd2b0ce0fd9f336b7588"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-22 19:11:12.000000000","tz":180},"committer":{"name":"Zelenevskii Vadim","email":"wortellen@gmail.com","date":"2024-04-08 06:50:51.000000000","tz":0},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/babb381bfe1a58be337aeed15c5e7ea6c248c3da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/babb381bfe1a58be337aeed15c5e7ea6c248c3da"}]},"branch":"refs/heads/master","description":"Rebase"},"cadfe9dca2490afc05d2c2489db700a417cd0f0d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-04-08 07:24:41.000000000","uploader":{"_account_id":36440,"name":"Zelenevskii Vadim","display_name":"Vadim","email":"wortellen@gmail.com","username":"wortellen"},"ref":"refs/changes/34/913334/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/913334/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/913334/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/913334/16"}}},"commit":{"parents":[{"commit":"9a011cb2779a766e22dc8f41736d68d3acc63d33","subject":"Merge \"reno: Update master for unmaintained/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9a011cb2779a766e22dc8f41736d68d3acc63d33"}]}],"author":{"name":"Vadim Zelenevsky","email":"wortellen@gmail.com","date":"2024-03-22 19:11:12.000000000","tz":180},"committer":{"name":"Zelenevskii Vadim","email":"wortellen@gmail.com","date":"2024-04-08 07:24:41.000000000","tz":0},"subject":"Fix maintence mode","message":"Fix maintence mode\n\nAdded MaintenceHook to app configuration\nRefactored maintence service\nFixed sql queries\nFix maintence controller\n\nCloses-Bug: #2058009\nSigned-off-by: Vadim Zelenevsky \u003cwortellen@gmail.com\u003e\nChange-Id: Ica41e8208b29aeba6e781801540843d46375c3d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cadfe9dca2490afc05d2c2489db700a417cd0f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cadfe9dca2490afc05d2c2489db700a417cd0f0d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
