)]}'
{"id":"openstack%2Fmistral~928638","triplet_id":"openstack%2Fmistral~master~I5e71586727f51c6ecf7da3019f66f68e502dd947","project":"openstack/mistral","branch":"master","topic":"release-health-check-cwrc-dalmatian","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-13 10:25:52.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I5e71586727f51c6ecf7da3019f66f68e502dd947","subject":"DNM: gate health test","status":"ABANDONED","created":"2024-09-09 15:26:34.000000000","updated":"2024-09-13 13:00:49.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"701ffa0b873c45bb690a5f52b03be58456c810fc","_number":928638,"virtual_id_number":928638,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-13 12:29:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 17:34:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"240561e6e818f4a0915e05d8e53cd9a65bcec2ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-09 15:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea466e760aaa17b86b32800776b097e00c69d161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:34:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87f778e9bb3041d0a21a279c94aec3ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d68f8f2d8b2c494d8a68786cdb30ac37 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b974d662be44cfdb3da12aed568eb2f : SUCCESS in 12m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/effc6b099aa64a90878454aa52f472b3 : SUCCESS in 14m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98de872b1a1f4656b9e0f2f02682829f : FAILURE in 11m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02ba106acd145f3b2808250fcee6ce3 : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4c4a10f35fe41cd89007e20e2e9a813 : SUCCESS in 18m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6da942a686c940e083bdb706e3d05311 : SUCCESS in 33m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76b82ef9dbea4629a96f76b97f506b38 : SUCCESS in 31m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/345b5a246f744d2696c509d315a8dc23 : SUCCESS in 37m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f5ae95704eb142c9859064b70d41de9f : SUCCESS in 31m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/738daf751a8c478a820bfe469cd8bdbb : SUCCESS in 40m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7f2f8707f8bd485c9cf8aafbd359c1ab : FAILURE in 46m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cce434f7aeea4e519d716f4a2e98ba17 : SUCCESS in 53m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8b0ac3cdc0ee482fb5161a7224ef8d43 : TIMED_OUT in 2h 04m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86a60999fae02e66cbb6ace086f9fa3d93a9c2c2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-13 10:25:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7e36f627b1bb89d3a177c8a2a78ae9ff668cb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 12:29:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8eb650ecb3943a9b09c9ff500c582b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf589b65c8ea45878e523b6a06669819 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8fa9aa4ddbe463cb78fa05a4bd2c3b3 : SUCCESS in 13m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3751a58e8ea24a94a4f1d8a6e0a85af6 : SUCCESS in 9m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5e06160e8704bfea5370b57d9021755 : FAILURE in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d9e0caf3b84f84add9ea96c441d77e : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb64ff04907b4c7ba6e2aca7329e339d : SUCCESS in 18m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9c2146b6bacf49b5b1a8b178f0296e56 : SUCCESS in 29m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/855e636bc6354b18be87458bb04e5e73 : SUCCESS in 25m 55s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/484724a25ac84b1bb99f78a5f94d0711 : SUCCESS in 25m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/af68e5c10b02422faf0d5f8e98923dc9 : SUCCESS in 38m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8736549a1ab24a4e9674d861787da85d : SUCCESS in 30m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/141f54130b774e8e862f464beb8601d8 : SUCCESS in 57m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9043e23dca164e8bbffb54645d3572a7 : SUCCESS in 53m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/73eb8faf03364bfb968aaa0fe2efbf18 : TIMED_OUT in 2h 02m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"701ffa0b873c45bb690a5f52b03be58456c810fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-13 13:00:49.000000000","message":"Abandoned\n\nOK, it seems that the error was just a temporary failure on gate, let\u0027s abandon this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bb6be23706b47a207b83671b94a3065a25e9ae2","revisions":{"1bb6be23706b47a207b83671b94a3065a25e9ae2":{"kind":"REWORK","_number":1,"created":"2024-09-09 15:26:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/38/928638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/38/928638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/928638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/928638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/38/928638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/38/928638/1"}}},"commit":{"parents":[{"commit":"a3e8f5b0437c2497b1f5e80e50fb873df38f9f53","subject":"Fix some race conditions in mistral CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3e8f5b0437c2497b1f5e80e50fb873df38f9f53"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:26:22.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:26:33.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I5e71586727f51c6ecf7da3019f66f68e502dd947\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bb6be23706b47a207b83671b94a3065a25e9ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1bb6be23706b47a207b83671b94a3065a25e9ae2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
