)]}'
{"id":"openstack%2Fmistral~929487","triplet_id":"openstack%2Fmistral~stable%2F2024.2~I29477e4b8d9a31e401497445f96e3585b106fa70","project":"openstack/mistral","branch":"stable/2024.2","topic":"create-2024.2","hashtags":[],"change_id":"I29477e4b8d9a31e401497445f96e3585b106fa70","subject":"Update .gitreview for stable/2024.2","status":"MERGED","created":"2024-09-16 12:31:34.000000000","updated":"2024-09-17 13:07:24.000000000","submitted":"2024-09-17 13:06:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929487-create-2024.2","meta_rev_id":"759d4a33fd56c53e4f3f421b611be69ea48f425e","_number":929487,"virtual_id_number":929487,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 13:06:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-09-17 12:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-09-17 12:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 13:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 08:55:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-09-17 12:08:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"dc43db53ade20f7174cdf125a60bb1520c9cdad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-09-16 12:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"190a9448d2c0f65497663426c9b74077787ba0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 13:32:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba80a5994fc14292ae1015a918ef6d4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057a12cf9ff242b3bf468751c8bd60a8 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51b85858876f4367bbf38520cb1bce55 : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cc25f1ed9aa4b82aa958b659b251dd3 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85fcfadf323f489096037ead12150001 : FAILURE in 12m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c25031845a649169a016c1d6951ab2c : SUCCESS in 7m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55058704df9c4fbdaeaad0e4408ac830 : FAILURE in 15m 50s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a6fe507aa8d443e684e870bca5021673 : SUCCESS in 44m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2fe7ed8645134ba8b9db41726fed3bbc : FAILURE in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ae81dbbadb8c3c12b565c89512083e6b66b511f4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-17 08:55:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d8d0eb14ee3591567b8418e4a4d8093cf8ccc3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 09:57:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca29ba8307294b228b82cbf400f4c1f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd25a807b9074cb690be5d9c221f80a9 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cb0dc77726c4f8395521b13dc45471b : SUCCESS in 12m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64cf494916a74dab8cc189e58bbd76ec : SUCCESS in 11m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2117de32c53d411d85dd145425526032 : FAILURE in 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3130425c5c48abb56703f645b9c283 : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/759c29aaa7c14ea29b06c72634cbceb4 : SUCCESS in 17m 30s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4c87a0f7cb144c2eb80167f14af31ed2 : SUCCESS in 52m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/43b1743e9f3b4e1cab5ae095dc714590 : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"94ef825d5b961ad1607248749244b3c568fe9fb1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-17 12:08:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0f0adb560e865619138983d646fae7bf2acbd15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 12:09:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35299008b4c3225ffbfb2e16166fbcb65d04d84f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 13:06:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5bdf725fe794ed0ab5a8312e69682e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2af330c5ed4dbd9d383260c5eef87d : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e19832cdbf004c2aaf37c12b37613b9e : SUCCESS in 12m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7624187218334372a40473bd5d6c0b0e : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0762a5f089f148539102cf0720ba60fb : SUCCESS in 5m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fabc8a3b33dc498ab377ab886d68cbed : SUCCESS in 52m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2f7077bf77f94bdba4991af8b2101d59 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c4fd5b6db46ffcb104ad1d0339edd4d4b6b66c20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 13:06:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"759d4a33fd56c53e4f3f421b611be69ea48f425e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 13:07:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/439622ac361545d8b8191f4c7a64ee67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7784f236abb04adebb7c22680e66d557 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90f220221a8e5bc213fbffa69ba7d7b4d500f933","revisions":{"90f220221a8e5bc213fbffa69ba7d7b4d500f933":{"kind":"REWORK","_number":1,"created":"2024-09-16 12:31:34.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/87/929487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/929487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/929487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/929487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/929487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/929487/1"}}},"commit":{"parents":[{"commit":"a3e8f5b0437c2497b1f5e80e50fb873df38f9f53","subject":"Fix some race conditions in mistral CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3e8f5b0437c2497b1f5e80e50fb873df38f9f53"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-16 12:31:33.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-09-16 12:31:33.000000000","tz":0},"subject":"Update .gitreview for stable/2024.2","message":"Update .gitreview for stable/2024.2\n\nChange-Id: I29477e4b8d9a31e401497445f96e3585b106fa70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90f220221a8e5bc213fbffa69ba7d7b4d500f933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/90f220221a8e5bc213fbffa69ba7d7b4d500f933"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
